)]}'
{"id":"openstack%2Fdesignate~888709","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I4c1b2829dd07ce0dd48c6ac6d56a52ecc1d83b98","project":"openstack/designate","branch":"stable/victoria","topic":"gate-fix-cap-virtualenv","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-08-01 01:29:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4c1b2829dd07ce0dd48c6ac6d56a52ecc1d83b98","subject":"[stable-only] Cap virtualenv/setuptools","status":"MERGED","created":"2023-07-17 15:12:49.000000000","updated":"2023-08-01 01:30:30.000000000","submitted":"2023-08-01 01:29:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888709-gate-fix-cap-virtualenv","meta_rev_id":"7ac24e61622076031e7fcd139a94e2a54d583f79","_number":888709,"virtual_id_number":888709,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-01 01:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":2,"date":"2023-07-31 16:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2023-08-01 00:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":1,"date":"2023-08-01 00:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-17 15:55:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 16:58:06.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2023-08-01 00:53:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"0130684fcc08dda9ca85452b25f6c6596085c956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-07-17 15:12:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78540dab1668da3aace1319ea04ae309636eba52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 15:55:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4bb35d5a1334f188feb506b4d897bfd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/492b74c9091942638b257e9667ae0e62 : SUCCESS in 31m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ae37758fe4804caaa164e518cacccd0b : SUCCESS in 29m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/46ad92f78a954ec2b5112e8423830236 : SUCCESS in 42m 27s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/95ecae78408644e6bbd0c1ee1f4d8574 : SUCCESS in 34m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/04724ffd00f642faaa0e064ecb0c12c7 : SUCCESS in 35m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c12c8913f0b5468695ea5f4dc4ebd5f7 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2162e0cf428f49d9b2f55b6984dadfe9 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6818553bd8684c4a933f15eb97b15e14 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fc9a0ffd9b6439aa4f563158eeec238 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3367ad69391c4a2eb3b295a330444192 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a78768b7df4348edb03672f3b981c1fe : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0cac8aaf672a46ee6db7440164ff5cd654d9703b","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2023-07-31 16:58:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e1285465c759d55b8911b816bd82a937652a1b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-01 00:53:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6985302275a081ffc6c540e6b4e4cebb6234cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 00:54:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"315761cb4200e36b77020f4dabd4fa2cc44f0b16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 01:29:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45ad5737454e4d85b5ef668fab7869d9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6ae481e710d148cd8f265c72e770bbc3 : SUCCESS in 33m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/375d8f19963b4d55a259911589320d48 : SUCCESS in 28m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2698f4b09e57498793556f9300bb5fe2 : SUCCESS in 30m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/288579d7647a441abce3277916e639be : SUCCESS in 32m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07819ffa9b664d1399c8c67c4d65980f : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37c7a0f502094481933baeb96e916c4b : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cfaa5154d1b4d9e9acc1982e8e9f2d8 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7824451502409caa99223f497318b5 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/075c5370801f436091270f52419d82f1 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"aa31980f499f7935df133711e9f2ce1f96ac42ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 01:29:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ac24e61622076031e7fcd139a94e2a54d583f79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 01:30:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0798ffd9eda24023ba3299d3ba62d388\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900af7bab0334c8ba919bc299b2c27c1 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be64160088904b05b1657cc30687086f : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4673e091fe9fcc1d6154f6430c74f8e4f56203e5","revisions":{"4673e091fe9fcc1d6154f6430c74f8e4f56203e5":{"kind":"REWORK","_number":1,"created":"2023-07-17 15:12:49.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/09/888709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/888709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/888709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/888709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/888709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/888709/1"}}},"commit":{"parents":[{"commit":"4f2c2a5c0ea68b1716899394f4c1249646807b55","subject":"Check zone_id in notification_handler to avoid UUID format issue.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f2c2a5c0ea68b1716899394f4c1249646807b55"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:12:31.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 15:12:31.000000000","tz":120},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: I4c1b2829dd07ce0dd48c6ac6d56a52ecc1d83b98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4673e091fe9fcc1d6154f6430c74f8e4f56203e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4673e091fe9fcc1d6154f6430c74f8e4f56203e5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
