)]}'
{"id":"openstack%2Fdesignate~888747","triplet_id":"openstack%2Fdesignate~master~Ibc0569235997ea1feb2341adc1e942ecb3fc6dcb","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-07-18 09:36:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc0569235997ea1feb2341adc1e942ecb3fc6dcb","subject":"Fix a sqlalchemy 2.x compatibility issue","status":"MERGED","created":"2023-07-18 01:32:14.000000000","updated":"2023-07-18 09:37:27.000000000","submitted":"2023-07-18 09:36:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888747","meta_rev_id":"6c5bcce2a67c84282cacc68020b96b407c361660","_number":888747,"virtual_id_number":888747,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-18 09:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-18 08:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-18 08:39:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-07-18 02:56:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-18 02:59:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 08:39:25.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"31ac031c8f70ecef514086d5d9b291b31fc4635f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-18 01:32:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5ce43675917215a26d413fe23578898a9ba8a96","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-18 02:56:03.000000000","message":"Patch Set 1: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52bcebab71edad0b672fa878e781fd040e8f8881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 02:59:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef48e200596f4e538c7b10ffc977238a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/05c4e53c7ab04fa7b3107f441260fecb : SUCCESS in 34m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fca3e7f11cfb403aaa4de03348e5d284 : SUCCESS in 40m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6edb1cd516674df1a4366d51327771d9 : SUCCESS in 48m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a6a58ff5f8ae4200aa5b5e37acc5289e : SUCCESS in 41m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e366a9313cb24b2ba5bdc46e6662f479 : SUCCESS in 52m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d513facf6fa448e6b08b4b61b15659a7 : SUCCESS in 1h 15m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5130a69b47a24f88a615ca0551de78fe : FAILURE in 1h 21m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cb87347e88864fcdbf1008f790f44ac0 : SUCCESS in 57m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/219b7c67e66645d0ba27c4ed84cc36c8 : SUCCESS in 50m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10d0f19f859c4cd49ef1da7a5475299a : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd911c28dd44b5db72bfb5bc49f36a8 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/591be3f867d04b438708d112159c884e : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea20f4d885a84adda00da8a5027d1c8e : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9478a82a94b44979bb0c12ad77e2d1b8 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d8c114858749fd822811e5e951dded : SUCCESS in 6m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/698cb0edac0b4f478633ef0d5f4ef233 : SUCCESS in 32m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/108253c5ce4148a887ffa7697f98e595 : SUCCESS in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efcfc2ae59d6c6c44ea19c9681ed2e35829f5182","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-18 05:52:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e853d7306064be73f82259b76ea0c2ae1f42f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 07:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc09bd9b0a1438a927acc2dfccdd611\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/79d5eb0fd4d94750b4a7ea934bb8e111 : SUCCESS in 35m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b363f45b542f496e9e1eac02f2eb970e : SUCCESS in 1h 08m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/190f8674eb8a4ed6869223c148ff3271 : SUCCESS in 50m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/23e99fb178934955bc75be5852e69706 : SUCCESS in 53m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d8b209d92a27472ebd3d85320f6f6d98 : SUCCESS in 56m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6e5e8b88226e4a1185a6382856777835 : SUCCESS in 44m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5821385745ce4f279544a960de2a9484 : SUCCESS in 52m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7871fc2e7e9a47c1b29762631e0d6261 : SUCCESS in 29m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e100d06dcae1435dbfedff4c33e8adb4 : SUCCESS in 31m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1adeaf2d7a4ab69a69a72e5a795c12 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3588f0210ebc4496bec56763d751e62e : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02988aec558b4d64b6906086b05f172d : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6818eb7aaecf47c6a54b8b6d5e555b75 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2eb87b9279847bbbe1e8979613a2d7f : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0148914fca2461e996e662898bcc78f : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8825bd97304846ffaf987b483d02b727 : SUCCESS in 32m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/03818ab683024a68bc5b679c81ae8027 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af21d938d552da66fe40cb5b052327e0f8266ce8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-18 08:39:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ad738818bf3e4bfb11b852ac7cd17c9bb6bce22b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 08:39:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a351c7d140d323b9b67a70417f872eccec7e074c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 09:36:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d3cf969a31e4ba3b99c09f73f85716c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e44e43a0bdee40fa9a80b92e23edc279 : SUCCESS in 49m 28s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c485368acc3425aa70b04c97f49feb9 : SUCCESS in 38m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/009887d3ce494aceb73a73859324562b : SUCCESS in 28m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6579f07762194b0cbca39b6e546c9d34 : SUCCESS in 50m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a0e5bc7e48d64527a7aac78056476349 : SUCCESS in 48m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7ca7b20614c6401dadcb1418662110de : SUCCESS in 40m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e07124f29cc429c82583d76f996df94 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f11d3755ad94b69a840266371977ec6 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5827522bb86544ada57fb35e6eb5549f : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a46749cf18134b57b84f9eadd635b94e : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a081d2ecb9f4533a7d80f36040d39a5 : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5658953c1ceb466e84043c21393d911c : SUCCESS in 42m 11s","accounts_in_message":[],"_revision_number":1},{"id":"894c42543ecd41b9ceb4acaff6684bd962bc623d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 09:36:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c5bcce2a67c84282cacc68020b96b407c361660","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 09:37:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85e921d58ea04a8fa779943870f2142f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa6d42a05ed4be39add0507bd2a110c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7daa2578a51cad61fed8e645d9e5158d5c5f5c17","revisions":{"7daa2578a51cad61fed8e645d9e5158d5c5f5c17":{"kind":"REWORK","_number":1,"created":"2023-07-18 01:32:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/47/888747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/47/888747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/888747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/888747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/47/888747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/47/888747/1"}}},"commit":{"parents":[{"commit":"963751f2b403c3e9a73008ac62854ed3d3758247","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/963751f2b403c3e9a73008ac62854ed3d3758247"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-18 01:30:02.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-18 01:32:13.000000000","tz":0},"subject":"Fix a sqlalchemy 2.x compatibility issue","message":"Fix a sqlalchemy 2.x compatibility issue\n\nA recent patch[1] introduced an incompatibility with sqlalchemy 2.x.\nUnfortunately the upper-constraint of oslo.db \u003c 13.0.0 blocks us from properly\ntesting with sqlalchemy 2.x, so this issue was not caught.\n\n[1] https://review.opendev.org/c/openstack/designate/+/880514\n\nChange-Id: Ibc0569235997ea1feb2341adc1e942ecb3fc6dcb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7daa2578a51cad61fed8e645d9e5158d5c5f5c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7daa2578a51cad61fed8e645d9e5158d5c5f5c17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
