)]}'
{"id":"openstack%2Fdesignate~889131","triplet_id":"openstack%2Fdesignate~master~I650089ad59ee87841540228a9c517b02f85f3677","project":"openstack/designate","branch":"master","topic":"tempest-job-timeout","hashtags":[],"change_id":"I650089ad59ee87841540228a9c517b02f85f3677","subject":"DNM: testing Tempest run increased concurrency","status":"ABANDONED","created":"2023-07-20 20:54:37.000000000","updated":"2023-08-02 23:35:16.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac9f0c4b9c5af3044a54883569a64ec211be05d3","_number":889131,"virtual_id_number":889131,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-20 22:00:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 22:00:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45a7504e7a70f32fc915a8899cc170264fd184c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 20:54:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0419c3475800d26d98ebdd8a01dbff859753b54e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 22:00:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e57340c76f584e4eb24e23648621f04a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d987e50ed542449b854d38bbeb43b02c : SUCCESS in 47m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3f1b0ceef15144bb9eb32b2a96f05599 : SUCCESS in 46m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d014f25b3fc4a2cb57a23580b39e336 : SUCCESS in 52m 26s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4a94bb72971c4ec0a432f4765e9fc839 : SUCCESS in 46m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a165e7cd9dc4e928eb8cf9450ffd328 : SUCCESS in 35m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/98b01f0b78174c069b79d8e635f1628f : SUCCESS in 54m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8b7c27d13acf4f7e8123e00adf0a1493 : SUCCESS in 1h 04m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5164fde9742f46d99a87ce4c1e609cee : SUCCESS in 36m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/77e427ebd81143579765e57ca661aac0 : SUCCESS in 49m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b52967b0f5ff48a2aeba684ff63c18c4 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9ce988d54b47d080eb85fddf12ca16 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ea707f4f8824b8fa81b35b57f48f76d : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76836d2e09344620866cb00483f3b27b : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a42db8a84d048c2b0745277deb60e83 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb501fbed314c49bc37109deebb038d : SUCCESS in 9m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efdb15059a2a4218b20b4e075698ac0f : SUCCESS in 35m 00s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e5b9854ca7504ce7bde4981a2d679f65 : SUCCESS in 7m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b3c07e7ae54c46f094a47c48b040a4ce : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0797c3c28ef212b92f1cdf2bf113c0436282979","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 23:14:52.000000000","message":"Topic timeout removed","accounts_in_message":[],"_revision_number":1},{"id":"2a737b942e991b70097f719b1b3086f73a06d4f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 23:14:56.000000000","message":"Topic set to tempest-job-timeout","accounts_in_message":[],"_revision_number":1},{"id":"67f3ce96f63c19a70cf32fd43acb588025102a83","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-24 17:05:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef4009dd80ff98cbb1bafa1cc6c7d37d5db9e503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 18:32:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d917b966851d43c7a32bad383484d457\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8f34fd19b51f46658e1b9a35aa24f5b2 : SUCCESS in 31m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5f6cb08a9d46432799e9d2f6422c25cb : FAILURE in 42m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5ba5c48593d442780138c1b3195143a : SUCCESS in 46m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/03fd38ca9a564bc28936fb7ed254e8d9 : SUCCESS in 34m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6b9853e38617410d85b9be83612d5cb5 : SUCCESS in 37m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e6b7b5e89d7a4728987c817b7696d848 : SUCCESS in 1h 06m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/005b4489f30b448e9cc21c32079543f0 : SUCCESS in 1h 15m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6ce8055ed5a848ce8df96206aeaf607d : SUCCESS in 41m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d4c02f5bafd54b75bbd25fcd622db844 : SUCCESS in 37m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f715ff804aa493b852644bcfe1b4e7f : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3d4e8d2699946e8b422791efeb14452 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d07e46b65774607b41ccb98c6d5dc54 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cd786affd2943139ebf613b5cd902c3 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df5b37558eb0483b8e390b85faadec04 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5d3069b0a134119843886c10d71ae3f : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/312316aaf09b4504ae75ec13b33ea49d : SUCCESS in 32m 26s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/09408579c171417c939197cea58caffe : SUCCESS in 7m 58s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/340939320bdf40cdb2fa52d8bd08a793 : SUCCESS in 54m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/48c10dd6dae0495ba1af83184ec4388e : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac9f0c4b9c5af3044a54883569a64ec211be05d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-02 23:35:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f33684c7b20b7505a1f96411209a33e60f519e2","revisions":{"6f33684c7b20b7505a1f96411209a33e60f519e2":{"kind":"REWORK","_number":1,"created":"2023-07-20 20:54:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/31/889131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/31/889131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/889131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/889131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/31/889131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/31/889131/1"}}},"commit":{"parents":[{"commit":"b77934beb7d2c707b2a2e10e023a2e2542c6cfe8","subject":"Add job to test with SQLAlchemy master (2.x)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b77934beb7d2c707b2a2e10e023a2e2542c6cfe8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-07-20 20:53:08.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-07-20 20:53:08.000000000","tz":-300},"subject":"DNM: testing Tempest run increased concurrency","message":"DNM: testing Tempest run increased concurrency\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/887220\nChange-Id: I650089ad59ee87841540228a9c517b02f85f3677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f33684c7b20b7505a1f96411209a33e60f519e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6f33684c7b20b7505a1f96411209a33e60f519e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
