)]}'
{"id":"openstack%2Fdesignate~889369","triplet_id":"openstack%2Fdesignate~master~I3a902c60b59071b466e1bf1ec6bda205e58b378c","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-07-31 19:57:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a902c60b59071b466e1bf1ec6bda205e58b378c","subject":"Replace blockdiag with graphviz","status":"MERGED","created":"2023-07-24 23:48:13.000000000","updated":"2023-07-31 19:58:59.000000000","submitted":"2023-07-31 19:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889369","meta_rev_id":"69d78933adea9187f676b0dbda08f5ab920f64a5","_number":889369,"virtual_id_number":889369,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-31 19:57:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-07-28 20:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-07-31 18:24:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 00:02:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-28 20:12:01.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"277300c1be014ffde4c578a0e3e0969b1ee45d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-24 23:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"39e2ec8c2da43a24a8dd804ae319ea48738e2e7c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-24 23:48:25.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2de5993aa1f9b282a74ff29154d6d973d356a79c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-24 23:51:41.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"673ac5deb36af1cca8464c4ca532bcc5fab43494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 00:02:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e184fc34a784adfbf608d38da62c7d9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d2d167d2e748bda7b6ffaa1e893b5c : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c2cb1069aab4c3cb9d6bff8cdbf8c23 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a47ee2dada1f8aaddc0e361f22e049bc3f931fba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-25 17:19:34.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"629bb5d1918411121ec009b85ccbcc343a4efb05","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-25 17:19:46.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ed81a7ebde4c386b910e178e578893c3f9864b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 18:36:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a348e705bf17443fa8249954b8d25b53\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/17eb3f36236c44b6b73deaa5a7bf7908 : SUCCESS in 8m 32s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/33a21328e7b04f4191308176b4002bf2 : SUCCESS in 50m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9496cc88ceaf43538491c74c9665bcfa : SUCCESS in 1h 00m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5e0485a2bc6417f904c13309e3609d0 : SUCCESS in 1h 00m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5632673a63b24f2c83128a52cd95f6bd : SUCCESS in 56m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/90e1b154b2ed403cae290b2b8340b10b : SUCCESS in 59m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/699a77487ee645d6975efdd4a8fb5654 : SUCCESS in 56m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f4a53268246244c5a6a78aaba1ff51d6 : SUCCESS in 1h 11m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d61fb533df0444f8a3476deec2d60bf4 : SUCCESS in 53m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47800df95386430c90b11ec61e0bbf70 : SUCCESS in 51m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b608027062046728e72c651d42ffbd9 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d972d6347766439fbeeb3cb6646007eb : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c0ecc7dede43c3a00dd5502da4ea7e : SUCCESS in 15m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d019d0fb4a2d4ebf9e2ccbe7b94d311b : SUCCESS in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53acd2b2696c41ce9338f6bf90378618 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9714f1dee545a0859c41df24ac8934 : SUCCESS in 17m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00a4092161f84ffda28e899adaf2daf5 : SUCCESS in 5m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/56223e57e238403e91f657dc67767c61 : SUCCESS in 33m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/db7640ac96254141a4fa56095c829ff7 : SUCCESS in 9m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/496d40008a75429e8b37549f18981dd9 : SUCCESS in 52m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c845acd7769843d2a493c55137f31a23 : SUCCESS in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7cfbfec09ba989cd2725a4a5b95a823d6442848e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-28 20:00:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6179872bb87f28130c017abc1cc812efaef30337","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-28 20:12:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"453c3a746913eb366177e41375438febc0957c99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 21:21:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96808a087fc54f88b5647a71032e9569\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/04ef5cdf2f9f4b0f9f804abfd54c2bea : SUCCESS in 3m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d13ec83516d94236bdd1091a143c086a : SUCCESS in 47m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a053ff70e8e040928ac74ccf85efc881 : TIMED_OUT in 1h 16m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/92fc2a0289d8442ebc35cf7dcb1a813c : SUCCESS in 46m 29s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ea0da31878f44ab0beb97e908bd3c53c : SUCCESS in 39m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be11a9a3022a430ca08b15fd4a07f138 : SUCCESS in 44m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/75c6a45fe7f3457284a67b71ce9228f8 : SUCCESS in 1h 08m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b0800ee31ff0416b9c9075773537e7ed : SUCCESS in 1h 13m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f6c04b511374a77bcc2120c33048a90 : SUCCESS in 52m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2c2d7189adc0459f844c93b63529d31d : SUCCESS in 39m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0cafc2c9b6a4936b758a178be08cb39 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f7b9b942b5422598687967c45f0e27 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3f18182b858425d95298a033be217f8 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3415b5a900fd4501be272af57cf7eb95 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29db7db8c449476592bfd451647850f1 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0844ed673c34105950da5996c0152f0 : SUCCESS in 8m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/43eb68b8ab2d4e88b02cf4928cf95aa2 : SUCCESS in 2m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40f790ae45d6427c8fa84e3842c6f6b5 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/756d83ed8a3249cba7a7315c10068bf9 : SUCCESS in 32m 26s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b6b2b599fcb143518cc65c05b803934b : SUCCESS in 7m 18s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7dc68e1cfbef4eaf9cc63d94342d1d6d : SUCCESS in 43m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8360e0d9ebb5403a80c70c4a4f911627 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ed169d2f380fa897200353aeb4445f5463388129","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-31 18:24:36.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f10f51e35024ceb60e8c9df25c4efddb83cc247d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 18:25:13.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"27ce9546374fd167997fac3785af1afdb3ce0eaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:57:52.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e74b7311127444bfa3489c50c1c41742\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a15a9de952774e20ab55d0b8fa2b1576 : SUCCESS in 3m 40s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/adc142751b374841adccc664f62a064e : SUCCESS in 31m 45s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d2eba5050fec46deab4ec72c498f80ad : SUCCESS in 30m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21d1bdb8d905408eb934f12daa85208e : SUCCESS in 35m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/092af62d855a4caaa5302ee1fd766062 : SUCCESS in 47m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/41c55a72173a411d8fede89947422934 : SUCCESS in 33m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5aa09b448d3b4ae3898e8f9ab2b518a7 : SUCCESS in 46m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99b9883319854cdb92f89d0dfe17ad62 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b993c33ea7ec4789891861404d4d04c2 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50cbf4d70d748e38bca99222eaa144a : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77d3934778ef4d53948dfaab2aeb11f3 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79eec6f5dd724b92b5c05c3580b3994f : SUCCESS in 8m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a7fcf07547341f8b18f4691ab703fab : SUCCESS in 1m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/054e85e2eda94380bea5f378158c3084 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/afdd0373f0fc4d4cac4e9716c1ba9bf4 : SUCCESS in 32m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4a217b09aba44043b8151bb42a507bdb : SUCCESS in 7m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/75c4bb1952974d21b0991edb4b1d380a : SUCCESS in 47m 21s","accounts_in_message":[],"_revision_number":4},{"id":"0628ff70ce261439653a1bf37c57a356d744ed02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:57:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"69d78933adea9187f676b0dbda08f5ab920f64a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 19:58:59.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08251815b4ee489fb1b06e1ce87a26ed\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/03bd5821633a45d18f8b625ad74edc9d : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db29b5a30e1f47aa96944c0d61079e30 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb5662a1ba774aa99332ea008c178a9d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"beb79ac98cf31f9c819fd73c512839dc22d39f4f","revisions":{"3535edf89ffa31db272da1973e7afa5557ac69be":{"kind":"REWORK","_number":1,"created":"2023-07-24 23:48:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/889369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/889369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/889369/1"}}},"commit":{"parents":[{"commit":"61b408287ff0bbb94e03f0fcd137af9f5413e2ca","subject":"Merge \"Add a tempest test job with sqlalchemy 2.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61b408287ff0bbb94e03f0fcd137af9f5413e2ca"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-24 23:46:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-24 23:48:11.000000000","tz":0},"subject":"Replace blockdiag with graphviz","message":"Replace blockdiag with graphviz\n\nThe python blockdiag module is unmaintained and will likely be removed from\ndistros.\nThis patch replaces blockdiag with graphviz diagrams.\n\nChange-Id: I3a902c60b59071b466e1bf1ec6bda205e58b378c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3535edf89ffa31db272da1973e7afa5557ac69be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3535edf89ffa31db272da1973e7afa5557ac69be"}]},"branch":"refs/heads/master"},"0e6177c272c4637bd681fcc5f8cda5b5151fce83":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-07-24 23:51:41.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/889369/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/889369/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/889369/2"}}},"commit":{"parents":[{"commit":"61b408287ff0bbb94e03f0fcd137af9f5413e2ca","subject":"Merge \"Add a tempest test job with sqlalchemy 2.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61b408287ff0bbb94e03f0fcd137af9f5413e2ca"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-24 23:46:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-24 23:51:15.000000000","tz":0},"subject":"Replace blockdiag with graphviz","message":"Replace blockdiag with graphviz\n\nThe python blockdiag module is unmaintained and will likely be removed from distros.[1]\nThis patch replaces blockdiag with graphviz diagrams.\n\n[1] https://github.com/blockdiag/blockdiag/pull/171\n\nChange-Id: I3a902c60b59071b466e1bf1ec6bda205e58b378c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e6177c272c4637bd681fcc5f8cda5b5151fce83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0e6177c272c4637bd681fcc5f8cda5b5151fce83"}]},"branch":"refs/heads/master"},"4e45fc973e7ba3ec17d5365c18295fc58d56654b":{"kind":"REWORK","_number":3,"created":"2023-07-25 17:19:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/889369/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/889369/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/889369/3"}}},"commit":{"parents":[{"commit":"61b408287ff0bbb94e03f0fcd137af9f5413e2ca","subject":"Merge \"Add a tempest test job with sqlalchemy 2.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61b408287ff0bbb94e03f0fcd137af9f5413e2ca"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-24 23:46:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-25 17:19:27.000000000","tz":0},"subject":"Replace blockdiag with graphviz","message":"Replace blockdiag with graphviz\n\nThe python blockdiag module is unmaintained and will likely be removed from distros.[1]\nThis patch replaces blockdiag with graphviz diagrams.\n\n[1] https://github.com/blockdiag/blockdiag/pull/171\n\nChange-Id: I3a902c60b59071b466e1bf1ec6bda205e58b378c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e45fc973e7ba3ec17d5365c18295fc58d56654b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e45fc973e7ba3ec17d5365c18295fc58d56654b"}]},"branch":"refs/heads/master"},"beb79ac98cf31f9c819fd73c512839dc22d39f4f":{"kind":"REWORK","_number":4,"created":"2023-07-28 20:00:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/69/889369/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/69/889369/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/69/889369/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/69/889369/4"}}},"commit":{"parents":[{"commit":"61b408287ff0bbb94e03f0fcd137af9f5413e2ca","subject":"Merge \"Add a tempest test job with sqlalchemy 2.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61b408287ff0bbb94e03f0fcd137af9f5413e2ca"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-24 23:46:23.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-07-28 20:00:07.000000000","tz":0},"subject":"Replace blockdiag with graphviz","message":"Replace blockdiag with graphviz\n\nThe python blockdiag module is unmaintained and will likely be removed from distros.[1]\nThis patch replaces blockdiag with graphviz diagrams.\n\n[1] https://github.com/blockdiag/blockdiag/pull/171\n\nChange-Id: I3a902c60b59071b466e1bf1ec6bda205e58b378c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/beb79ac98cf31f9c819fd73c512839dc22d39f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/beb79ac98cf31f9c819fd73c512839dc22d39f4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
