)]}'
{"id":"openstack%2Fdesignate~889828","triplet_id":"openstack%2Fdesignate~master~I03d98e2ab950247fad73814e2ac03a96cda749b6","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-07-31 15:48:40.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I03d98e2ab950247fad73814e2ac03a96cda749b6","subject":"Added unit test coverage for verify_project_id","status":"MERGED","created":"2023-07-29 20:12:27.000000000","updated":"2023-07-31 16:55:06.000000000","submitted":"2023-07-31 16:54:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889828","meta_rev_id":"a361fedce1a41022c8641b49fc5ba71a3bcc469e","_number":889828,"virtual_id_number":889828,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-31 16:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-07-31 15:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-07-31 15:48:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-29 21:07:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-31 15:32:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"89f170855495578dfa346ba63247f3e72c919036","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-29 20:12:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69adf07509d97c597191c5ef02c9db637786c39a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-29 21:07:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3892f6d3a0a4c14a9b4cb33be2622b8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6c6099a49e214510a493842f86180ff1 : FAILURE in 54m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f42c17e6237740eb82b1197c74f69612 : FAILURE in 46m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a732d7306024e6498f41238b0e8f55b : FAILURE in 40m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/039f591508d94a4a9324d1d95e50a11e : FAILURE in 47m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8cdf78f0f086485ea50f6516c4166b3f : FAILURE in 31m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d48e170204dc479b9349b4c6d708c790 : FAILURE in 43m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0630a805b0c14f53a3b531a0b2c166ed : FAILURE in 43m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2ec303d3686b4282916234f9f6da56df : FAILURE in 31m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc9a4e147b2c46ffa38b6f60c4963187 : FAILURE in 31m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b542d1bf76a54cdf93e9a7d1d1ef3bea : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e47eb0a8b31f47eabaf6d36cc1518fc6 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc135210f18a41f4abcf6fb68a65438b : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab3a1fad63a04766ad0599664e6f349a : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/455268b745ea4adaaef4ab5bde6a6224 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0c18e01db354513958d014383de858d : SUCCESS in 6m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f7624f9491af4fd3ae236f18171ba462 : SUCCESS in 32m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9c959f26827c492c8afe0e0765b6fecb : SUCCESS in 6m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be203710832f4e858b4b995d69d5dec9 : FAILURE in 28m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f10de9cd957c482c98735974483ea6d2 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da02680981777bd962ba371cdda751ef195dc517","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-29 21:11:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"8fa172da257a02f9e11d19b9a65e83a87bfad3e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-29 21:49:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a6aeec4381ff06727eebac5a1026022671046a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-29 22:48:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afce2fbd087549308355a6f605721f63\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/999717ec04df49e88819c7902c17e02f : SUCCESS in 53m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/afaa8fc68c67486bb3ef81e05fd89752 : FAILURE in 41m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aa2263f663d4e70b5dc194d717c7b97 : FAILURE in 23m 35s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/023e601df08f454fbe89d2f6fa95e3e2 : SUCCESS in 30m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8a902449a6e84050830b14bc5f3f5958 : SUCCESS in 32m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d6f24f8ac08c40d5bce7737bc2d8876a : SUCCESS in 47m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c689d2f835a4495283b960fe745051ed : SUCCESS in 47m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/06ddf53779c34d229e866be84c46acc3 : SUCCESS in 31m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bfe7f01fe1d84718a88637ffb3fa858e : SUCCESS in 33m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2166f6dca90c4502a68c289b8774c0b9 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d419631c032c45789d4e9fe07739f1f8 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a952638063fc417cb52e647653510a77 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1bd59e0631243798820a79a0de217d1 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fef849df5a684f08a9139c21b254efcc : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7068813c1f946f1bdf64b49fd36463b : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b0558dc1f0e4e3e8674780cdc686e31 : SUCCESS in 31m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e24e5157b084be8ad07b7e9b3a9b602 : SUCCESS in 5m 21s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/85afec7f5db34c6e9b3565b63f3aafca : SUCCESS in 48m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8bd1a0079e6c416bbed3c32b7c9480f1 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0cb1f1a944ee12370358716464a401ee3352a63a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-30 15:29:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2d5fc99ea1313a26d158fd85e98c5ca95e3fab15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-30 16:44:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdd3cc3b23e64c58b23fd72a919b42ca\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5f91a7be25e2496ab7c6627454cacadb : SUCCESS in 38m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/471df1f717784d1490e29549ab96866b : FAILURE in 25m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c30962b56bc6431f8da7cb0041435731 : FAILURE in 35m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2ee2083fcfe3483bb7736ebaa4f1e516 : SUCCESS in 35m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/077a7fda93b04cb4960fc36e92f0a378 : SUCCESS in 45m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/914158b466914568b6f183ec5b7e32e9 : SUCCESS in 1h 07m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9559279623514f1e8edf32dee2395d2d : SUCCESS in 1h 00m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/eb345371f055452fa303013af6e4116a : SUCCESS in 32m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eef2adf815a6432db326e2f9b9fc9447 : SUCCESS in 49m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a834a5f2d3b347ddbba70dfed0bac82d : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394f42872cb140d4ad4bae479d1d3dab : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87395f30845c4476bc455df7724046dd : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df107c5b15054ea1b1338ac0f671432a : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4aca942f37f747c99eac5182d7d2d5f2 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556fd03cc677486b904367141825b2e0 : SUCCESS in 9m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd32f717170641f8b5f1195020e9b593 : SUCCESS in 34m 30s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1bd4a4cecf3b49e1b8746095a098896a : SUCCESS in 6m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5fc8e23a36894698b799decc46dc7f57 : SUCCESS in 46m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9a227118783b4a6b87a993afbe16b66c : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2427f44e76ad5b0934eb04b41c63efdd86c6702a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-07-31 15:32:29.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a28543b3f62d28cb9dfc8d2f70927839f6c65333","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-07-31 15:48:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e4b94d2813183e03e446d269badcaab4cd270623","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 15:49:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"54a03e18eb5216a119302e8ee551426da22f2089","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:54:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/111cb9f6fed94dbeb36608cf68e6e904\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32d1af2df30d40f8b4e2aeba4d62e6ba : SUCCESS in 38m 35s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7de70bdcd5854065bd01c0aee06dfcf1 : SUCCESS in 38m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/67b5160d9dc04669b3b475a4c11bfc06 : SUCCESS in 37m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4657cf643f32412590b45c502ba13af5 : SUCCESS in 1h 00m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ddc3eb4dd5ea4095bc379dded07b2cfd : SUCCESS in 30m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/286dc9278dbb4fe5ae51848ecefe6bc3 : SUCCESS in 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d667b98b809443cba6a3ec56fae9bd31 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/079d4b952b1345f981cdc1ec9dc75091 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6886ca4dc086493ab1183824ef47b1c7 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e2a4ce2f0924af3ae375faad9930917 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd46a30de4384150925d6233ecef11b7 : SUCCESS in 6m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2458e4cea06946e890976f7363c934a7 : SUCCESS in 30m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9033c47c7f094d6296dd59a7d11b322a : SUCCESS in 6m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d9998c7a690b4aa5bd088ae32879df9a : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":3},{"id":"e13949a4a03d792a17c5104f7e8835ed62a20b49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:54:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a361fedce1a41022c8641b49fc5ba71a3bcc469e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-31 16:55:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/980fd59d94144132aadeee608e8c4fc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae3e33f351a41e29acfd6c506a10645 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f371b1af0269aaf3ad3de02f0ac65044a653733e","revisions":{"3592855e3b867bb3a406384a4b5743905eca51c6":{"kind":"REWORK","_number":1,"created":"2023-07-29 20:12:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/28/889828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/28/889828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/28/889828/1"}}},"commit":{"parents":[{"commit":"fa95018c919957ce730e825a11073e840a6af616","subject":"Cleanup DNS Middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa95018c919957ce730e825a11073e840a6af616"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-29 20:11:43.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-29 20:11:43.000000000","tz":120},"subject":"Fixed potential issues with verify_project_id","message":"Fixed potential issues with verify_project_id\n\nFixed potential issues with verify_project_id and\nadded test coverage.\n\nChange-Id: I03d98e2ab950247fad73814e2ac03a96cda749b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3592855e3b867bb3a406384a4b5743905eca51c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3592855e3b867bb3a406384a4b5743905eca51c6"}]},"branch":"refs/heads/master"},"0a45755a0f093fa1b2654528d13ad226a411e38c":{"kind":"REWORK","_number":2,"created":"2023-07-29 21:49:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/28/889828/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/28/889828/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/28/889828/2"}}},"commit":{"parents":[{"commit":"fa95018c919957ce730e825a11073e840a6af616","subject":"Cleanup DNS Middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa95018c919957ce730e825a11073e840a6af616"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-29 20:11:43.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-29 21:49:12.000000000","tz":120},"subject":"Fixed potential issues with verify_project_id","message":"Fixed potential issues with verify_project_id\n\nFixed potential issues with verify_project_id and\nadded test coverage.\n\nChange-Id: I03d98e2ab950247fad73814e2ac03a96cda749b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a45755a0f093fa1b2654528d13ad226a411e38c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a45755a0f093fa1b2654528d13ad226a411e38c"}]},"branch":"refs/heads/master"},"f371b1af0269aaf3ad3de02f0ac65044a653733e":{"kind":"REWORK","_number":3,"created":"2023-07-30 15:29:22.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/28/889828/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/28/889828/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/28/889828/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/28/889828/3"}}},"commit":{"parents":[{"commit":"fa95018c919957ce730e825a11073e840a6af616","subject":"Cleanup DNS Middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa95018c919957ce730e825a11073e840a6af616"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-29 20:11:43.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-07-30 15:28:07.000000000","tz":120},"subject":"Added unit test coverage for verify_project_id","message":"Added unit test coverage for verify_project_id\n\n- Cleaned up code path to make it easier to read.\n- Added unit test coverage.\n\nChange-Id: I03d98e2ab950247fad73814e2ac03a96cda749b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f371b1af0269aaf3ad3de02f0ac65044a653733e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f371b1af0269aaf3ad3de02f0ac65044a653733e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
