)]}'
{"id":"openstack%2Fdesignate~890073","triplet_id":"openstack%2Fdesignate~master~I3ca7c4b5ae4bba736bfa16333f87b341e36be8db","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-02 19:32:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I3ca7c4b5ae4bba736bfa16333f87b341e36be8db","subject":"Add basic SRV record unit test","status":"MERGED","created":"2023-08-01 21:40:48.000000000","updated":"2023-08-02 20:37:52.000000000","submitted":"2023-08-02 20:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890073","meta_rev_id":"4c5369d8f14fef89cc4ffaa4bc6f7f090505a7ac","_number":890073,"virtual_id_number":890073,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 20:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-08-02 17:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-02 19:32:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 22:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 17:57:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"144e32eb6344f26734038a243a89229c6698837a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-01 21:40:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75a657f038164fa3df02350e017212d33bc986cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 22:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abbe1ef25f254586968f670fdd4274e3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/02583c326b7044c982127a1aebf48dfe : SUCCESS in 36m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b535b1504b54361a9f3f21a0c0831ec : FAILURE in 31m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0caee12fffc940afaed2f31c3b9631bf : FAILURE in 21m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a16038a38e0340acb32a9d56e9ff2258 : SUCCESS in 49m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/233749e10e86498c8115553bb9d3571d : SUCCESS in 46m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fe0a3b1223694c40ac13d9468782e7b3 : SUCCESS in 1h 12m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f458d666daed4f8190e1185ad76251ed : SUCCESS in 1h 10m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8e6470e897da4b8a81039e947603e33e : SUCCESS in 45m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4cc9a21aca6c41b48a428861af8563d4 : SUCCESS in 55m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a66260c9809947ce90c1bf0bfa4bae38 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cbad10b0e8d45718d03a63e572cc6c6 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dacbf4e3a5644cddbb78fa0192d70b32 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d87687f538b4f4cb6c44a1cfe29e605 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15f0e7c7879a42dc9d18e89fc1375189 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1575d4165455487983a9ff951893fe81 : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fc25ea62a23e4b18a9933252e3392c71 : SUCCESS in 40m 04s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ce5aa795e43f46f48104c25cfd966a4b : SUCCESS in 6m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/146beacfebf04b36980c2faa67d3c34e : SUCCESS in 41m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2e01e2fa4eaf4c538e1aac9e7c1ace18 : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84ac97259e5ce11b3d4d2b210e47520d65376e8f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-02 17:57:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8210a38dedd32e89db274690fdb77070f5ef82a6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 19:32:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54ecf27c3c7e56a66ed71fb6046b03a1dfb13b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 19:32:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce517fb854ca0e2c0fe646f5617f174608307341","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:36:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab604bc6dfd49e2abbf6bac1ec7bcf9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/105e5c8eaee044038f5208ef34578384 : SUCCESS in 49m 19s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1cddb6c689b940329fcea1f4dd1e3f23 : SUCCESS in 36m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eb7bc945108c49f5bad2e3851028ad5e : SUCCESS in 41m 39s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/03a15aa60a5c4b89b36ebaf528d2d614 : SUCCESS in 1h 02m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be0399dd390d49d6a936efb8de7e76be : SUCCESS in 37m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b25f3450bdee405aa4fd7a2d5cbe0445 : SUCCESS in 54m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9553491e8b2d455180f4b06b94025093 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7de33911b0c437bb0c11fbb6411029d : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7438451b3dba453090625b50055a272e : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed8da2a562fc4bbeb1baf60ca0284c25 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87823714adea455e9b4c64b730cea9d4 : SUCCESS in 14m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/943edf28d41e40228efb1c4cbf0ac4ba : SUCCESS in 33m 04s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/44fffbfb5e4641f2854da7ee0fb1342c : SUCCESS in 6m 13s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4e5d40dac9854308844ea87aef315b5a : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":1},{"id":"36924670b816d90beb257f26820191414fccc044","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:36:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c5369d8f14fef89cc4ffaa4bc6f7f090505a7ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:37:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f40f53a19f5a4c54bb885079674098a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f68b791a8ab4e3eb68f09f6d03c6e50 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"167a01fdd9225b678c37213b1f0309b0f5cd886f","revisions":{"167a01fdd9225b678c37213b1f0309b0f5cd886f":{"kind":"REWORK","_number":1,"created":"2023-08-01 21:40:48.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/73/890073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/890073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/890073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/890073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/890073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/890073/1"}}},"commit":{"parents":[{"commit":"025854ab88132b2ab0ac6131ca50f95c57ccd7e5","subject":"Merge \"Use i18n and not gettext for infoblox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/025854ab88132b2ab0ac6131ca50f95c57ccd7e5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:40:33.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:40:33.000000000","tz":120},"subject":"Add basic SRV record unit test","message":"Add basic SRV record unit test\n\nChange-Id: I3ca7c4b5ae4bba736bfa16333f87b341e36be8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/167a01fdd9225b678c37213b1f0309b0f5cd886f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/167a01fdd9225b678c37213b1f0309b0f5cd886f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
