)]}'
{"id":"openstack%2Fdesignate~890075","triplet_id":"openstack%2Fdesignate~master~I3cb738f2775f770deda30253b91f148b593314bc","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-08-21 23:37:39.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-22 10:38:18.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I3cb738f2775f770deda30253b91f148b593314bc","subject":"Moved all regexes to constants","status":"MERGED","created":"2023-08-01 21:57:41.000000000","updated":"2023-08-22 11:56:29.000000000","submitted":"2023-08-22 11:55:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890075","meta_rev_id":"bd13a39eade14d57a3e5f7cea2afb8c21e220a48","_number":890075,"virtual_id_number":890075,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 11:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-21 23:37:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-08-22 10:38:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-01 22:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 16:14:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1edc1ad89587f04a9194212583a6c8fc551e7707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-01 21:57:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13c0a5ef11965d3c10db4599d32f2592753f6725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-01 22:51:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b468dc1c395d4209b2560ab947935c37\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/05f2625c1ff94fa882ecf70bd8c998bc : SUCCESS in 48m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/803d6a501a9c4ff68a7aec6ec702b1d3 : FAILURE in 40m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdd9bd8ab24f484d9fc4163611a5641d : FAILURE in 37m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0e6558c4eb404cad818154c8c803c915 : SUCCESS in 48m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2a5b1fc5f0c6495cbdd35277dc0e3a9e : SUCCESS in 32m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c3a37262979f4af7b456042421d0e43f : SUCCESS in 51m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8daa360a6624427f9a333e698d12bba4 : SUCCESS in 49m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5e32452507384efd91a818078a068243 : SUCCESS in 34m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/24c9ddb202b44ccfa5d27fceabe0dc64 : SUCCESS in 33m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0e640571025419abb80809e0aee66d7 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24670fadc67143f5880ef270fb4ff118 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/083abc108c9348c3958f40cc13779de3 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c5be5ff56bd404b8091084fcae0f6ba : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c272ece7e144c0e9071053e0b110d98 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a947f0b46a84e57b9518efeef60523e : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/75324b75faae43d5b96335d4032940ed : SUCCESS in 30m 55s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/42feebf320ad400da75552f8219ba430 : SUCCESS in 6m 54s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e2721d7452f14ef1992c2040d118980d : SUCCESS in 32m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9d1aa4f325a248609512de0638076fd8 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4577437e728c678d040a2d28af1c7a1b222eec51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 14:45:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6020b4707778edca8f8dffcc1225cd0541825239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 15:24:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9440d06796db824015a117a429bd8cf29fc7b9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 17:13:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b71b0db1c39c463db2d0317802f824be\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8df82ed225574628aa658bf00429b7ca : SUCCESS in 39m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8f53e7f455d2486ba2dddf729febce36 : FAILURE in 24m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/66ecd0c13b8f47fe9b0b0e41c03db9ca : FAILURE in 23m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1d906694c5054bf8a1f08ca6082f1981 : SUCCESS in 36m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7efb0bba0afe4eac9fbc39c319034b03 : SUCCESS in 35m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f972ecc9b04f46d28bcaa41e0cc861f8 : SUCCESS in 1h 10m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f29f52b8058a4e2b87ec16a6adfa5b32 : SUCCESS in 49m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/066080093e9e477595fd8d0ea1e97cf6 : SUCCESS in 34m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a4a9bd5794ee44348abf3c5c8e3ac6dd : FAILURE in 39m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f16bd2928154a76b9513e3ad8f76f11 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4766a5db586431998e1afff09d723ba : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cf2748cd2ab481884fd335232522d0e : FAILURE in 7m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2bafd9e7e7640f1913781cb82f4c243 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a9066965cc346aab4c31fa91829f13d : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b7e9ede2fb4ef4869d4000e0dce3cd : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdab467f1f1b4175875c0cab5b373f95 : SUCCESS in 31m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4c3d2e0f25b349abb59cd46b24247abf : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8e10f172da3746f6ad39ea12fa665889 : SUCCESS in 49m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/222436b45c70434585a84434fe9a6fb9 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0906dcb3c52d20f0139751ba2284a4cc0ea99801","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 19:29:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b7dca112e9b3e56406921b2fa01fe3ec54da0b82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 20:47:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7f66429b8ec85456d03ef6e8e7ef29525eaac701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 21:09:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8fc8481e02834825b9fefd43fa8521c70bfebfc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 22:07:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cff2aaae6204f7eae8deaca77adcb6e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40319887a8a847c4ab33ac6cf8f35040 : SUCCESS in 29m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eba9d58eae3a4d7fb2c75869f0d22ec5 : FAILURE in 29m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0926e66b55ea4114a0d6002bc7ca0395 : FAILURE in 35m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ad77cccb4abb4d85acf81a59cfd5d3f3 : SUCCESS in 47m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/52dbffc9c66d4113867acc9222bec3da : SUCCESS in 30m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/744c1148eccc406b92c46e281e2f7bef : SUCCESS in 53m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b9c8dfc9bfb4773a3a6425f48fcc77a : SUCCESS in 53m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2731913ddeca47369f548e5ef2c26ebf : SUCCESS in 49m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c3b2176a3c66433495106527ec404094 : SUCCESS in 35m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49b7aecae5ec4661974c5b73a6d82939 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed120afffd8e4c7098a799c9b94a8621 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6260552f56a942f5a0e225c5a0eccbe7 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a285a281c6c94a82ba4b968d507474f0 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f877be8ebf15497da2b203b3f0bd70dd : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfe4887a96a425383c60d8869aa0ee2 : SUCCESS in 9m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/62ae1c7f218b479b92ff410524ad3647 : SUCCESS in 32m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9df58db06c2040978e19648866f8e764 : SUCCESS in 6m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0fc8c9f171145d39ffa40cc9bfde7ed : SUCCESS in 48m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2862721713154ddda4daef9749ea1be4 : SUCCESS in 4m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fc22ce7e58654baf6a838e5f709f746d2656d174","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 23:04:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1270d64b7b1445cddad6a0a1aa4f07fee262a219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 00:21:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feddec4bf6004f1f9035537a2a1a2b49\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/91d1b2a81f79409ebe29ebdcdad9f0f8 : SUCCESS in 50m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/471bfc2dfa0545eda56c2f5cc9f6a8aa : FAILURE in 29m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcddddf5f0d748a988787603f2d3d7f2 : FAILURE in 23m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2e201e6215ea47ecb3919f2f8fc6a45a : SUCCESS in 36m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d71216a62a524be2ac653b06add0e738 : SUCCESS in 46m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/39b508d306914355851bebc201623cbe : SUCCESS in 52m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ce32323991844a0a95393f5ca1886984 : SUCCESS in 1h 10m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8d8a4d233e5a4edd91c3ad4d2893b7b8 : SUCCESS in 32m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8dd1427ad6b24de686f93a0e0540aa6a : SUCCESS in 59m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8a71ab3f96c4d998ba265c82a4c2ae6 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbdca16f62504c998f009cf9698aca02 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f65469865d35403980636407b4bca293 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f634f247526147b1a3556614e3873363 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f49cb65a9ad4f018a1803a14c8dcbb4 : SUCCESS in 9m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c403675d6de344ec8bfc7660ad6a4b63 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1eaf14dd95ae492e8e4952a40a697851 : SUCCESS in 38m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cab4dcf672674919a2ca0a38489ce132 : SUCCESS in 6m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6203539ca90e494e9920d5c43c3a3597 : SUCCESS in 32m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9da40e890c784c348af13e9e7e2531f0 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3edea48c9dea50b37ff394242c0e3f145e0730ee","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-03 00:36:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d71dd8f2a3e62d2db4318e552d3361920c20f99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 01:33:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c2fcc2b2d494c7d80558584ada8f9d9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e76d0d7fdc64a21bf77198267fd6736 : SUCCESS in 37m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/be2211d1a5da4e9f91dbbba2a1b72a5c : FAILURE in 29m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0a96ff12b8eb45f4b8295a4d962fdbb7 : FAILURE in 20m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d6363a61373d4dc099aad1def9c2acb3 : SUCCESS in 34m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3c0c28fd6e3044c79b25071999c3aefc : SUCCESS in 32m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3ba98ddc05f749a3be12e9c90522b1bf : SUCCESS in 55m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9be4362b6b847e7b7449bead1344b9e : SUCCESS in 52m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cd38991c78834e0e99f535926d184d65 : SUCCESS in 26m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/55ea437f4e3d4e019e3c5a4e10fddab9 : SUCCESS in 38m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816675dc7d724471936a265e3fd5a5c7 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c013ba741b245ed821364833de664f6 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25c3f48a931e4bc2863654510e4889cc : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a726929aad014e08841ff2767fb35d8f : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e64e9b084e94a3e8b30f3ec4f7b954b : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ab3394ac0243349a3eb7ca19ca0e79 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/96f5ebeab982459183249093ee651691 : SUCCESS in 34m 10s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/904987df39c94f72afcee723bdcaee3e : SUCCESS in 5m 51s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/88ba9b99599e431c998c15d9b8d24e36 : SUCCESS in 37m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7b2cf7fb961540249df5b43bf7537762 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6bb91ccd85ef431e3ae419113da5f1c7386426d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-04 16:14:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7e29acace62e19cd9f905ae3ea5071b22527c6d9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-04 16:24:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0844ebced7950732de250ee83751100378173d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-05 15:19:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9b099ca21d697bf9ef15d75d62c804322aa45569","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-05 15:19:23.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"31038ef73a28172b7917c565f8c0151f0acff7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-05 16:34:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5f017acf06a4737955829b1f5a43346\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/145efcd18c844a51b1132eca0a7e66e3 : FAILURE in 38m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/20af86647d7a4495b768715f8be33f32 : SUCCESS in 1h 02m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d8fcfbe524d47038b07840cf9a97dfc : SUCCESS in 38m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9b988e1d736b4b3e8e43686f6bdc6bbe : SUCCESS in 35m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f91582a643a34d5381032389673f4f0e : SUCCESS in 36m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/59ebaa892f0848babd21b7720c411d7a : SUCCESS in 57m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7cbff41a2bf41249630f6e36966d230 : SUCCESS in 1h 10m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1032b63cdd19457a8d9560646860ee43 : SUCCESS in 50m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/87e6d473ad434c4f9a41542e170d9ec3 : SUCCESS in 48m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fc338a31510402293633771ea6d7170 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7447f45d759c4799baa47a33b7b629cf : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08b4afa08bb34f2c9d901300a147822a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7166f0dcb2d84143bb6a16286fa06930 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0d4126eb3cb48daba603bd86243d975 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ab21dbd0464001b175808c6c96af56 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8dee9d3eb2154eceb11feb39fba25522 : SUCCESS in 33m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2ea3756dca42499f8996cd66880ccb08 : SUCCESS in 5m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8e48a02193b8427b9b400a36fbce7273 : SUCCESS in 38m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a1d4ddeaff7d4119a9b4e50e5a77a266 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9f8d01df4ba3acfc05fd7b1ad77d6d438824ae44","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-05 20:37:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6f71f1c3b3b750433b6aaafb59ec2dffe1a12872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-05 21:50:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f37bcbd1b4864d039bf2232cb512843f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2b7de30e91ed47e4a76bbd8e9e3b8307 : SUCCESS in 56m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1a7a261c2d0d43378c34a0aa46b3dae6 : SUCCESS in 1h 01m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bdd4aa59b4340cabe100e2eb0ca18df : SUCCESS in 54m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c2bf0ebd103c4d29884ad1f6ff96afba : SUCCESS in 32m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a304202be4743bf932fc612df553183 : SUCCESS in 51m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/16186b33ea924e9487694be526b14a89 : SUCCESS in 1h 07m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/90067e45861b47f3ada912a2e00f0f34 : SUCCESS in 41m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8fa693fc21274fc7872a218c361d2179 : SUCCESS in 47m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a58084b95c52472bb24ddf86bccce628 : SUCCESS in 31m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26882e620ef6442e9dc4a13c4b434790 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4a750b511ae474e87307e3f1da1a552 : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b032dffe0dc1440b987bf5099299d655 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/520a14ed02b4454cacd45992176ae4ea : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c096b2fe0b04844b5ccc960b3926836 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d0e3ff0a864a19bb192c3c6cd3d922 : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc3444aaba5e49f0b786251f92b701d1 : SUCCESS in 34m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1cf10ebf59c140e2bad5de567647940e : SUCCESS in 6m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a9ce7f9a9f6d420e911339c064fe1998 : SUCCESS in 50m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c3504e3d0b9644eb87b5209a4f48b180 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"39d4aaa017847fa9245b44e52973fef38eac6a88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-10 11:56:57.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e6bbc0a88d6fe51ed560091b8f30ad74df50fed6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-10 11:57:03.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"9948a5e7ea5d37df9b83b83a54e727eb9763e4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 13:19:26.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21a0266ff8aa4fc1bc59d54db415cc79\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7a06daf99c464f138720ca05b639c5f6 : SUCCESS in 36m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8cc0b143c1594a01b0b5297f7b0b51fc : SUCCESS in 47m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/db8d0d5e04a74e9a9c8474bd45528cf9 : SUCCESS in 44m 31s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eefc1b74300b4a649b76405c035a4e77 : SUCCESS in 52m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bbc18051cd4f4b5aa2fa06a91cf2a513 : SUCCESS in 37m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f5559c44456644f9ac7d1d32097c708e : SUCCESS in 1h 14m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8dea082dd594c08aa05c63ad147a440 : SUCCESS in 57m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/08fdf3b3a647409f9ed3afc99c4badda : SUCCESS in 43m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f71c864b56e45c397414b27e5430637 : SUCCESS in 36m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc4d15806ca4cc3bfa437088e93813c : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da604f8e96a844c590e6b55a2caf90d7 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9ab8bc8558e4d9d92b4752b20f5d505 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6faeff0fded844ee94e0d0a548db9865 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55605402b9814eb5b5ecc915eb959b0d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e57a8e8d4094c5ba32596b5672c1e12 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5bf50633e1484a29acaf4d6df7d88932 : SUCCESS in 34m 13s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7246a58194b143de95f594224bb4112b : SUCCESS in 7m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/18f5296e6c48429d90ab94de24588ff1 : SUCCESS in 42m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/504fdbb4d6d442719a32d39b265da337 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a07e36c1086b580c9ce8ba663f986a5a9cbf9255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-14 17:21:24.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"cb620058afccea8a9599ae969437469e188c8fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 18:20:51.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f927e7bf2fc542aaa47d70de0fad48d6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0231b9ce5b84448f896892a7ea40f598 : SUCCESS in 48m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4374b652c7374ec6b9e9f8e6bddc03d0 : FAILURE in 33m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6683fe62d164bbc9575da6aa45c0f70 : FAILURE in 30m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1e8a037d6e1b4caa87e1cf089de7d6ba : SUCCESS in 47m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/01fe08ae4aaf4e6d90d981923e47c9e2 : SUCCESS in 38m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a52def74eed84892a3a2570460ef670d : SUCCESS in 57m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9b1f3f4e41cf4852bb403fb67a76d54a : SUCCESS in 47m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d3c874c680984fd88a9ff466282ac436 : SUCCESS in 39m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e5602079be84e999607e5755a4cbd01 : SUCCESS in 36m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b97a8ae32f449e7aaf542be3d8c8acc : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf7c92ae6437445681645e4ad451fcae : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcc10a8f0f7d481b8ccc92806f47e66f : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0874f97d57845eba9265e668363d129 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/711d8d12f89443688cf05618ce95e54b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4299d661db52421eb08c25707bdc599e : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3f5f3b0a0d894c4eb5ad11d26e9dd623 : SUCCESS in 41m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a08e1ef3e5ee4e8e993f60950cb7eeaf : SUCCESS in 7m 02s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/82f6ea4bf3a744499dcd1cefce70964e : SUCCESS in 29m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/82f49b228b944466986895368bc56abe : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7fef5285827ccefa92acc5ce2d0a193ea511316c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-21 23:37:39.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"46e3bca7cf0e39bbfd4510987b1fe0521050ecab","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-22 10:38:18.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"7a6776290516215f21eec7875416cd64a40a029a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 10:38:55.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"8ffc0345e87062d6fab41aab1418f735a58d6e99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:55:13.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7acd7af3b7434dd7ba0cb7e86f6063c3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2ad46474cf8d486fa84287da268f8b01 : SUCCESS in 34m 17s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8240ff98bbda4dd1ba09217539508a1d : SUCCESS in 49m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/822c95d4155f402cb8f162df74b18183 : SUCCESS in 48m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bb5f5ada024048cb8bfb12cf84cfc09c : SUCCESS in 1h 11m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f8b341577d5d4a25a53d4e8291818040 : SUCCESS in 36m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bcc4a93315ea4ca0b929e64738f8399f : SUCCESS in 44m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/799a3bb459b04b8d97f881d4df4d36d0 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/265aab8d8d194e16b82d3ba946b5daac : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c5d51cb548a4825a3290e86a8ed8c6b : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04924729810b47048cebec0bfc534d26 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf5284db7ef46d09fdbb04bdb571d43 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e74ab5ab7801473db2d751ac5030ce87 : SUCCESS in 43m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3526d2d9e73643deb61cd40fdfc5f4f5 : SUCCESS in 7m 31s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e0d24d93bb5c4d64bcad89d89d23d69f : SUCCESS in 46m 13s","accounts_in_message":[],"_revision_number":9},{"id":"cb311a77268f9a9010009653779710d8652ec176","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:55:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"bd13a39eade14d57a3e5f7cea2afb8c21e220a48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:56:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75dd3d8af6a4cd281210569aa78125d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e839f7c91f14554b87dfc47559ac5ac : SUCCESS in 54s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"18a5d4f4d9e6400fd1be97695b253d1acf920555","revisions":{"5a4e0f0f9cc82356baccec949fe2b414565bd68f":{"kind":"REWORK","_number":1,"created":"2023-08-01 21:57:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/1"}}},"commit":{"parents":[{"commit":"c8828b0728f78141d0e6c20917d9c42cabc419a4","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8828b0728f78141d0e6c20917d9c42cabc419a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"subject":"Move record regexes to constants","message":"Move record regexes to constants\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a4e0f0f9cc82356baccec949fe2b414565bd68f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a4e0f0f9cc82356baccec949fe2b414565bd68f"}]},"branch":"refs/heads/master"},"e745f17b9a3ba659a8da7afbffe500c0bb96eebe":{"kind":"REWORK","_number":2,"created":"2023-08-02 14:45:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/2"}}},"commit":{"parents":[{"commit":"c8828b0728f78141d0e6c20917d9c42cabc419a4","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8828b0728f78141d0e6c20917d9c42cabc419a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 14:45:23.000000000","tz":120},"subject":"Move record regexes to constants","message":"Move record regexes to constants\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e745f17b9a3ba659a8da7afbffe500c0bb96eebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e745f17b9a3ba659a8da7afbffe500c0bb96eebe"}]},"branch":"refs/heads/master"},"39a388d8d8c3a4e77f555e6d92e74da02b654328":{"kind":"REWORK","_number":3,"created":"2023-08-02 15:24:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/3"}}},"commit":{"parents":[{"commit":"c8828b0728f78141d0e6c20917d9c42cabc419a4","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c8828b0728f78141d0e6c20917d9c42cabc419a4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 15:24:05.000000000","tz":120},"subject":"Move record regexes to constants","message":"Move record regexes to constants\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39a388d8d8c3a4e77f555e6d92e74da02b654328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39a388d8d8c3a4e77f555e6d92e74da02b654328"}]},"branch":"refs/heads/master"},"5e9bcdab000fceef633b9b88a91c9b8f347e3155":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-02 20:47:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/4"}}},"commit":{"parents":[{"commit":"3c5d03d4023a1804d9a2dd4abe2dedda956807d2","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3c5d03d4023a1804d9a2dd4abe2dedda956807d2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 20:47:32.000000000","tz":0},"subject":"Move record regexes to constants","message":"Move record regexes to constants\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e9bcdab000fceef633b9b88a91c9b8f347e3155"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5e9bcdab000fceef633b9b88a91c9b8f347e3155"}]},"branch":"refs/heads/master","description":"Rebase"},"b35d41f34d917a191efc4aa8083cdf14c5cbe5e1":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-08-02 21:09:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/5"}}},"commit":{"parents":[{"commit":"5c9fca9d428b2ab7d6e24bcce5ea9190a1f4fe2e","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c9fca9d428b2ab7d6e24bcce5ea9190a1f4fe2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 21:09:28.000000000","tz":0},"subject":"Move record regexes to constants","message":"Move record regexes to constants\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b35d41f34d917a191efc4aa8083cdf14c5cbe5e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b35d41f34d917a191efc4aa8083cdf14c5cbe5e1"}]},"branch":"refs/heads/master","description":"Rebase"},"598c85d174139e4d77187106752a63f0d1f812e5":{"kind":"REWORK","_number":6,"created":"2023-08-05 15:19:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/6"}}},"commit":{"parents":[{"commit":"5c9fca9d428b2ab7d6e24bcce5ea9190a1f4fe2e","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c9fca9d428b2ab7d6e24bcce5ea9190a1f4fe2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-05 15:18:14.000000000","tz":120},"subject":"Moved all regexes to constants","message":"Moved all regexes to constants\n\n- Moved record regexes to constants.\n- Moved floating ip regex to constants.\n- Moved valdiation regex to constants.\n- Added some additional floating ip tests.\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/598c85d174139e4d77187106752a63f0d1f812e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/598c85d174139e4d77187106752a63f0d1f812e5"}]},"branch":"refs/heads/master"},"7a205299a965ab21b8a28e9b76741c81386efad9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-08-10 11:56:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/7"}}},"commit":{"parents":[{"commit":"5c9fca9d428b2ab7d6e24bcce5ea9190a1f4fe2e","subject":"Removed unused object fields code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c9fca9d428b2ab7d6e24bcce5ea9190a1f4fe2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-10 11:56:57.000000000","tz":0},"subject":"Moved all regexes to constants","message":"Moved all regexes to constants\n\n- Moved record regexes to constants.\n- Moved floating ip regex to constants.\n- Moved valdiation regex to constants.\n- Added additional floating ip tests around regexes.\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a205299a965ab21b8a28e9b76741c81386efad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a205299a965ab21b8a28e9b76741c81386efad9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"74e9d0a0b86783ae959a1e736aba97827535b43f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-08-10 11:57:03.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/8"}}},"commit":{"parents":[{"commit":"5fa2dd8efc65723d68b2ae369e86e85f9ab2e48a","subject":"Merge \"Fix missing oslo.versionedobjects library option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5fa2dd8efc65723d68b2ae369e86e85f9ab2e48a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-10 11:57:03.000000000","tz":0},"subject":"Moved all regexes to constants","message":"Moved all regexes to constants\n\n- Moved record regexes to constants.\n- Moved floating ip regex to constants.\n- Moved valdiation regex to constants.\n- Added additional floating ip tests around regexes.\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74e9d0a0b86783ae959a1e736aba97827535b43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/74e9d0a0b86783ae959a1e736aba97827535b43f"}]},"branch":"refs/heads/master","description":"Rebase"},"18a5d4f4d9e6400fd1be97695b253d1acf920555":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-08-14 17:21:24.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/75/890075/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/890075/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/890075/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/890075/9"}}},"commit":{"parents":[{"commit":"29097ed55953964557f28e75858bd211ce1c7bca","subject":"Additional recordset validation testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29097ed55953964557f28e75858bd211ce1c7bca"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-01 21:57:20.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-14 17:21:24.000000000","tz":0},"subject":"Moved all regexes to constants","message":"Moved all regexes to constants\n\n- Moved record regexes to constants.\n- Moved floating ip regex to constants.\n- Moved valdiation regex to constants.\n- Added additional floating ip tests around regexes.\n\nChange-Id: I3cb738f2775f770deda30253b91f148b593314bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a5d4f4d9e6400fd1be97695b253d1acf920555"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a5d4f4d9e6400fd1be97695b253d1acf920555"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
