)]}'
{"id":"openstack%2Fdesignate~890283","triplet_id":"openstack%2Fdesignate~master~Ic7f7bec89cb17b6c0ddbdbc93237ef273f17deb6","project":"openstack/designate","branch":"master","topic":"coverage","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-02 20:47:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic7f7bec89cb17b6c0ddbdbc93237ef273f17deb6","subject":"Increase minimum test coverage to 90%","status":"MERGED","created":"2023-08-02 14:52:39.000000000","updated":"2023-08-02 20:48:41.000000000","submitted":"2023-08-02 20:47:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890283-coverage","meta_rev_id":"193ac0f1441a8229656f132a85c5fb40c51ea69b","_number":890283,"virtual_id_number":890283,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 20:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-08-02 19:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-08-02 19:22:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 16:02:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 19:22:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"75593ecd54e0132a62f1a74f1d6b2859874df71d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 14:52:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cc9fe7f8b1fc672e8baf4ee4bf2a922e3f7e23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 16:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64fbdaeb12eb4d55bd1d2287a7ce0f79\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/28ce687c46874c9a8201a05de2e0aaa7 : SUCCESS in 4m 12s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d31596904d504288b08dcf22efadfe07 : SUCCESS in 1h 01m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/447a1b7b51fa48278ba870830a39f52c : FAILURE in 41m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/50d078e2267f4117b1bcec364cf48b99 : FAILURE in 48m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c5a5699a23bc4dc3acd2548d140b72f9 : SUCCESS in 52m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/84c2d849f4764225a7ac88ce8f7698be : SUCCESS in 34m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/027f0697e7a44ef298e65c713eb7eb19 : SUCCESS in 53m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e020a199e759474ea4ec11d7c89959c3 : SUCCESS in 58m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3f339cd408e24a2a812ff6cff3bf1023 : SUCCESS in 50m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/79a35a81c5f042b8b87625980e99695b : SUCCESS in 41m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d984af31f0c46ffa743cd36f0530b8d : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66926e8b363a4f5780f704f6384897b1 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da1d885cdb8c457eb31191c2919a2563 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dea002911d04a6ebef0d7717c98bb1b : SUCCESS in 7m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e8fb80029724472da3780c4665d47a33 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed22566e5c544ee8c8628278effdfed : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ffb1da8a1464e8ea70f4815e1c2a372 : SUCCESS in 4m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/46bf4411c2834e8ab241ebbb53d2b077 : SUCCESS in 7m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/46b5dfb409a247ebb99ee2597d7f8681 : SUCCESS in 45m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/380e20bd436f42b6be373615ac669e68 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68d937b7e8fa05c86e90728d72495964a13112c7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-02 19:22:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"386318b4e2f5ccd8f36dae0ea0ce750a0566ad14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 19:32:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4804cdee749d223c11cf440161fbd09d7b24c427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:47:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9acf136e5648a58541ced44c7a5e2c\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/728322c28ad348fdb2c5e97b904dd532 : SUCCESS in 7m 46s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f14be3bbd8d2498fa8774ed5eca44193 : SUCCESS in 35m 35s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1d3ff6e193e44a0fbfa3a64761b99666 : SUCCESS in 30m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/436cedc8625b414e8b10132434f40cb6 : SUCCESS in 36m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6cbb1c87d8874f68b65b4757f116a47d : SUCCESS in 1h 07m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/199bd1a4b0da4511931b19a08a224387 : SUCCESS in 50m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b7aaafa8fd974d25817b3d2b18af6742 : SUCCESS in 49m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f007f4285345a6b76c07ed0d897fce : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a64fd337b05140de962744e021f92b4f : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54b594d878274df591fe9d914d04b067 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/551a71da8f144e75a51835d35b13c608 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a8f800bb9845269ad2797a8b85ecab : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fada3594091447faecd4242caef22d9 : SUCCESS in 4m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6fc9282c7151423aaeb4acc617a9e269 : SUCCESS in 6m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f6d80a09ac4d4cc9ab9cae56b6cbabd7 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d554d26fd90d623378e3b1e0f01639091490c55e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:47:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"193ac0f1441a8229656f132a85c5fb40c51ea69b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 20:48:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea4b8606d5e04bbfaa58e8a9430faf0b\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a50a3768d8cf4a15bda5dd11b5ba5c5a : SUCCESS in 48s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5f15448a3a4de78ff041975a59ec0b : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d1aac21dcb44c13acc245d00481d444 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c102c0830bd4ab0e261dd14f9bc14d945d277f91","revisions":{"c102c0830bd4ab0e261dd14f9bc14d945d277f91":{"kind":"REWORK","_number":1,"created":"2023-08-02 14:52:39.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/83/890283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/890283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/890283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/890283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/890283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/890283/1"}}},"commit":{"parents":[{"commit":"167a01fdd9225b678c37213b1f0309b0f5cd886f","subject":"Add basic SRV record unit test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/167a01fdd9225b678c37213b1f0309b0f5cd886f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 14:51:37.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 14:51:37.000000000","tz":120},"subject":"Increase minimum test coverage to 90%","message":"Increase minimum test coverage to 90%\n\nChange-Id: Ic7f7bec89cb17b6c0ddbdbc93237ef273f17deb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c102c0830bd4ab0e261dd14f9bc14d945d277f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c102c0830bd4ab0e261dd14f9bc14d945d277f91"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
