)]}'
{"id":"openstack%2Fdesignate~890284","triplet_id":"openstack%2Fdesignate~master~I1a17a24f644c8dc28c4744baf7e05f4e011c47f7","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-04 20:57:39.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I1a17a24f644c8dc28c4744baf7e05f4e011c47f7","subject":"Simplify test_purge_zones to fix race condition","status":"MERGED","created":"2023-08-02 20:27:18.000000000","updated":"2023-08-04 23:11:30.000000000","submitted":"2023-08-04 23:10:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890284","meta_rev_id":"891e15a281d068599faa6acbe8af4389b5cbcd62","_number":890284,"virtual_id_number":890284,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-04 23:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-08-04 16:05:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-08-04 16:05:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-02 23:05:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-02 21:31:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-04 16:05:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"598696391f31194d068773f08a49736cd0896c41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 20:27:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19a5c96dc7ef48b183ba6f89db02600605db6df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 21:31:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82f15c8fa2b242a382f1e59e9b4fc0b1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae982f226d774bae8a6da3561e1c0c29 : SUCCESS in 47m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/aa006d17abe34b52899fb3632ec9305e : FAILURE in 38m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2875d018f67b4ed585279fe34b75cf0c : FAILURE in 35m 51s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f9fcd2d8d1d9411493bc3d8f0eda5211 : SUCCESS in 48m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5789499b3aef4d58a63b9a9fd3695fec : SUCCESS in 56m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8c954cce4b63478db4a25c1ddea78409 : SUCCESS in 48m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cbb4753f09b34ff99a77a6a51dfaf109 : SUCCESS in 49m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0a2c7e69223b4dae8433b6feb1732b7d : SUCCESS in 28m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0527ab8ba58144e1bb17784bb8a242a7 : SUCCESS in 40m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2310fdab9864f519462185de009a8c0 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dcc527fce7546cfac05c5368aeccdfd : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a5e2f5731284eaeacb95d04dd4b599a : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab3fbb4c82464bc29179a7114b1880ce : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00d10df79b114eb2bc1d2ea92bbcb6ba : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c5edd14a5f4fc9b6e15ff4f5eb48ef : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cbea7d80ee27469d9590512696a69ba8 : SUCCESS in 32m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e9cf5f8c69f74657b3e6651963373cda : SUCCESS in 5m 46s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a43cbcbb48c0452f95b62717d8f0661f : SUCCESS in 37m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/39bfdc23bb97490ca1640201c169a871 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"11be8e24d7aac138f56d8761968fb66deba6923a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-02 23:05:21.000000000","message":"Patch Set 1: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"509d3312c22d1b39e1f783fca0189f7dbd802ef9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-04 16:05:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3a9e5b6b9b420d2c243f230b59d0f0dd5caae3c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:06:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"585f8a9c1c2e600faf50ea3dab793b680221bdec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 16:38:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9efb5a355718439290c7e1711332164b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6caef50f299148cbb21b7726c577b7b9 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/96183a53570f451aa2c3f676f91b35a0 : FAILURE in 31m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/19cc180174a24b278a84fa23d44cae12 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0817d6dbda00416b93ec56f0cfea2f25 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6fe0d4bff48d44a580acefa42280abe2 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0476843110c14c5ebe59b8975d4ee020 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d8479bc69d4f429b27fe1165fc7931 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f793829094964d2783ffbda7cff53053 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c674afbc4c04d98b4cb34fa0e2369e0 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bde1f9cb879741b9a6350debfaef4682 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/716afa0c33844770bdb900644952f022 : SUCCESS in 8m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/38784f882e32443eb10b128f6dc454b9 : SUCCESS in 31m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7bf50e31ebee4f959b4b6e0394dfbfe4 : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8c9c726f0cf54f0dadd6f447b3373ab5 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"2264d788941894b9b35e028dffbd36ade2315661","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-04 19:10:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90cda69292fe2d8c25130bbaea0d1d87502e9822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 20:33:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c315544dc88f45e5986e4acd0e88db48\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e724caf199c4039ab7276d60a868460 : FAILURE in 41m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a6e8a1e5a3674f93b41a8d96550e9a49 : FAILURE in 31m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ae8edecd7cc4ef1839dd5c4ee4682de : FAILURE in 35m 30s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e05e994641e84b99b108b31ef938c239 : SUCCESS in 49m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/180b1fb78fc44678a9a5618dad0c2da9 : SUCCESS in 34m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/815d2defb0bd4b05b11c48934ac9ec8a : SUCCESS in 1h 14m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9da7b0a0438748f484491d0c19507ed7 : SUCCESS in 1h 03m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/66ba6f40537849edaafd0f6e52a41425 : SUCCESS in 31m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf076bf94adb45799e0e472e9683c990 : SUCCESS in 39m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e82ce74b2d340f9bec376b1b074afa1 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba84210407842a4b1c4a053454fe2e6 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40ca3d39e85441a680f1828a3f9a57fc : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c803150c95f94ceb87000ae654e786c4 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eae424ddbc3d4c848c9dc6d2a545e70a : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1563fd09566c46b39897d5ed5dcbea79 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7d4b8dbff124c8c8b9f17da25fb651a : SUCCESS in 32m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2e78a98c6d9a4c47bb9bf5ec26331cb9 : SUCCESS in 8m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/407ded26f5f548edb0b5dfa925140d3b : SUCCESS in 46m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/510cd2ac25bf40aa8fb2f7d261f14ba2 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"99f6250bfb9d6c695da9d564f41ad2b570eeb0fb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-04 20:57:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"029941460bc4deba0a7ce12a79091cbc91b61878","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 22:14:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86abd701373941859afbdb0360506dc5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb9230e412574230984d79e6b76bcb6b : SUCCESS in 38m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fa1bd0fc294945e2a4b01a38bd8b1002 : FAILURE in 27m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/11b7ee1a92f1483c904412d3ce54f642 : FAILURE in 23m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5a685accdebc47d9826765b81fefb217 : SUCCESS in 36m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9a26830e658340f690ce350a3b5dd451 : SUCCESS in 46m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c95cd657bc1b42679a30423a075f8b4a : SUCCESS in 52m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cec3058271a44312925ae309a61c6093 : SUCCESS in 1h 11m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4592aa772b474bcf8e62ffa634ae7201 : SUCCESS in 35m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/00a86c7e6a54472d9981f5a0ada6dedd : SUCCESS in 39m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d8b10aff4ef411ebac39aa6e19cc42d : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1c438b6e434b4f83279b2bae6cb357 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57050310ec774785ada9983f99be524e : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab03242f05b74679bb1c50764c7ebf70 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fec0be92af9a41d78dc7cd6d23c3eb29 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dbdcac01c8c44f0bcf6e1a9e8eb76bd : SUCCESS in 8m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51c11cfde09c4db1afbba943433e6e40 : SUCCESS in 32m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/db95b7bbe1c645329be48aacd0367942 : SUCCESS in 6m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e8fc4421324c4d22999ca1b47503070b : SUCCESS in 48m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/109b7c832cf648038f417eaa3e276e42 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2882ba94e5fd7f7e5bfa572899a8d56bf566c83c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 22:15:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"38550c4f851afc70d1fc439c7638026955248639","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 23:10:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca7f8bdbbac4885933a0efb7d2bc8b5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ccd5b30865c948e99ec3e49fb12aba27 : SUCCESS in 48m 12s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9b27bac27162478b86de5a1f24ba45cb : SUCCESS in 36m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f9b333f8dc8843598b062df5d97b6971 : SUCCESS in 34m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/61fb75faece84d039a4cebde0ca23c97 : SUCCESS in 53m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/00f257b39e204e17944a34bcee3d7c8a : SUCCESS in 37m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/68371fa11a6c4c4a91e214d3a7aef7f3 : SUCCESS in 50m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e82225dc034b43b289b6c85b8137edf9 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c94b196707e141a4a5e4d8c5a56b084a : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11aca561b6714745a609c23fda9ca558 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03c0099c9a694f31adb377fef1f46a70 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c022b5efc829400490f32f738daae39d : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6b3139fb987417283614af1449a8e56 : SUCCESS in 31m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2d96c7029ee14d2e99ec83e441b5a730 : SUCCESS in 5m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0abd76976458482aae395e4c275543cf : SUCCESS in 49m 17s","accounts_in_message":[],"_revision_number":1},{"id":"678dbb4792060fb8dba21a22a52c00a3bcf4a690","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 23:10:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"891e15a281d068599faa6acbe8af4389b5cbcd62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 23:11:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dd02bc59dc74af6812e6e7cc39bbcbf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eab9ce62ed54c5bb4e71ffa0f202958 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88d589115a436c551fe314f94e2ca699e36b3ee8","revisions":{"88d589115a436c551fe314f94e2ca699e36b3ee8":{"kind":"REWORK","_number":1,"created":"2023-08-02 20:27:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/84/890284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/84/890284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/890284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/890284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/84/890284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/84/890284/1"}}},"commit":{"parents":[{"commit":"031eb8b5d3a5acc39d18d0f9cc350545bd83fb7e","subject":"Fix indentation of user documentation code blocks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/031eb8b5d3a5acc39d18d0f9cc350545bd83fb7e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 20:26:31.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-02 20:26:31.000000000","tz":120},"subject":"Simplify test_purge_zones to fix race condition","message":"Simplify test_purge_zones to fix race condition\n\nThis is one of the tests that pretty frequently\nfails due to a race condition. Simplify it a bit\nto remove the risk of a race condition.\n\nChange-Id: I1a17a24f644c8dc28c4744baf7e05f4e011c47f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88d589115a436c551fe314f94e2ca699e36b3ee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88d589115a436c551fe314f94e2ca699e36b3ee8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
