)]}'
{"id":"openstack%2Fdesignate~890582","triplet_id":"openstack%2Fdesignate~master~Ibceb63090230aaf9348627e23ae55ca46764b9d5","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-02 20:20:52.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-27 22:35:03.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2025-06-02 21:43:27.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-06-02 21:43:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibceb63090230aaf9348627e23ae55ca46764b9d5","subject":"tox: Remove basepython","status":"MERGED","created":"2023-08-05 07:26:59.000000000","updated":"2025-06-02 21:44:31.000000000","submitted":"2025-06-02 21:43:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890582","meta_rev_id":"4ab51a0c61f0c5924d5f097a75565315dc4e6443","_number":890582,"virtual_id_number":890582,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 21:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-06-02 20:20:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2025-02-23 12:12:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-27 22:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-02-23 12:24:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-06-02 20:20:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-05 08:52:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-23 08:58:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-23 08:58:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-02-23 08:58:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-02-23 12:09:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-02-23 12:10:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"},{"updated":"2025-02-23 12:24:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-27 22:56:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"939a73a0343bc706834e7753cf89bd5c480970dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-05 07:26:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"193cf170fd95759c5dd4d8a689de513589ed6203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-05 08:52:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32abda478f4a431aafe45144972e5aa2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7184394025a143e88b007f56810fc0e3 : SUCCESS in 4m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7042e3e1aba74f9cadca33daa08bf6c6 : SUCCESS in 48m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eeb251c4b75b43f3a0b0a6f63eadc245 : FAILURE in 23m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0225807a22648988ba3cdc6b9702ce3 : FAILURE in 35m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f8a654caa7944304a0a964d98d371140 : SUCCESS in 31m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ab83277f6772481b9f14c54abb19ea78 : SUCCESS in 32m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eedbebc764164068b234d10da4afcb8f : SUCCESS in 1h 06m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/069556bdcd4041569751f228710ae1fd : SUCCESS in 1h 19m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c1d0c1318764bd8923f6e0f17e42323 : SUCCESS in 55m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/11f867354c9a44a2bf9c905fdc925dce : SUCCESS in 30m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2c2e44697b244f68100736a4f04048f : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff8856eec1864f1ab09760eb51bf51aa : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c1876404f649ffa88cf432d71eb896 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e062837cecc04385a4b86fc4494ebf99 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0acae30ea884e22ad3f5ec80d98abee : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12d5f0c2de394ab1b30cecc22b1cb8ad : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8a78463eb894cb4bb2a07891158492a : SUCCESS in 3m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bed3af34b78143fd9e84ad0e3431f582 : SUCCESS in 7m 20s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/70385ebde2e64fb0ac8f008164c426dd : SUCCESS in 53m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6f0cc83ba41c4a12adda188edcb0eab7 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cda6b74105cb7f456a371cb2a4e5323d74cb6c7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-22 21:41:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_32029\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"_revision_number":2},{"id":"41950e3c427e1bf25a94f8de7634488d38139bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 23:00:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97b1f83652b6466da8f79a480dad9127\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9c1f52ebb46344f6af3a778e380793d8 : SUCCESS in 3m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1790bb9631c0454abe817f403b157401 : SUCCESS in 42m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/593c2cd143ac49b8ae6e46ef002c6528 : SUCCESS in 53m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e991872473094ea8b362e7b3bf5856e3 : SUCCESS in 56m 49s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/7d222d15d0b94ba598d4563d2d2cd319 : SUCCESS in 47m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b529351d8e9340cdb5ff4ab68e75cff1 : SUCCESS in 55m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a6e41948d53744a1b5ca21749822bc4e : SUCCESS in 1h 03m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fb13a2b9a43147d4ba9d35869febbfe2 : SUCCESS in 1h 12m 44s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0b7565f3a1c0470fb07dc40227883da5 : SUCCESS in 1h 11m 43s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a7408d4a6f1b40e4a999e5b649b3ffa1 : SUCCESS in 1h 03m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5e21b22488ab48ec946794ed455be7af : SUCCESS in 50m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d02bf03f2ecb4bef83dc00a61887407c : SUCCESS in 50m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/174990ca72c74eb49c45d3007a52c704 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6b87dd83fc46e795dd70a952956ed7 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc2e28bbac2043428cfa1b4a62a470c0 : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5391e15f2d8c48d2a21ac6088fd3ed1b : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbfb32a091c4716b0124fc0face3449 : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/596f90201c404d6da33efd6ef4e19142 : SUCCESS in 2m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d67fae83225844a2b7d80b470f726433 : SUCCESS in 4m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ec45d3e1ee2844a9bf071a14b67f1365 : FAILURE in 42m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f198f9e0fdfd151909daca4cb774202b9fa7cbb5","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-23 08:58:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"144e4add91d0b6a79af35c424ff9397ddd7c850c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-23 12:09:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"effb55824fd583a2c08b0b8ffceec64e29f38dee","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-23 12:10:53.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"a78380d563d4c80729d90932aecc0fd8c379a613","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-23 12:12:52.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8bf668cc8eab0157b82123f156ad1c7ce92641a3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-23 12:24:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0fc5f4f91f046bed2bdb2dd505d3130267d08099","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-24 14:29:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db71c88c556e0b874de24181610dead282fcd0d4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-02-24 14:32:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6374afce42cf4165741175f43aa371d285c3bdc","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-24 15:29:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb335c93e71ed2b21dd31cc284e0dcf9f8598a60","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-27 22:35:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83ff27b91c71f3bbd63abeb8197ce9bc1d364ee5","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-01 12:33:37.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf39e5837ec698f85d7815d1153b01cdfcbde012","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-31 21:33:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"530a2f8e7d85d0b3244c4fe203dd0e699c6b2b99","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 20:20:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d501969169e097aac748140b9c95e50cba68fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:21:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4c69747122f55a00e7594a2f9f2f99fa53dd47e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:43:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b806fba4cec4fb98258bb7c641ee81b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/05104861f8004a3f987f633e6dfc317e : SUCCESS in 4m 19s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fbc9b631ff694b5db7a8baf67e130f8f : SUCCESS in 41m 11s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/66be2208d49147b39e680d656804c102 : SUCCESS in 42m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a3a0f8d991cb449a8815cde125fabceb : SUCCESS in 39m 10s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9f231ba4ada84f4a989137872dec12f7 : SUCCESS in 1h 07m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8089cd09b72240adaa28e52477e0666a : SUCCESS in 50m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a940f7f2d6f94d508e310064db43d62b : SUCCESS in 44m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bde8efec86c4e8c9012772ea265c90c : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a9c9d6dbec84a77bcefdb9ccc673e88 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/12d2b0bb91bc45ef9f9a10306d180530 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca66867905d348a5a23851e62bb7b1d5 : SUCCESS in 9m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/838f23e7aeb64d5e99feb9a429a8b110 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2f3c4ee4627870f7d1d9bca13081b7571f587346","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:43:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ab51a0c61f0c5924d5f097a75565315dc4e6443","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:44:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fd55301f4e2453d83d6ac0e264e42f3\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/59c25d8eba194efab70605047049caaf : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69fb8b327cc4b25bca90b91a2bd1117 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5fa7d1581fc4e98b9703038e2f18b8a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd693fcb926d5520ac7d51e343e12ceff96fa40f","revisions":{"d1fbd7f4843644c3dd4a548c5c19af4130be5528":{"kind":"REWORK","_number":1,"created":"2023-08-05 07:26:59.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/82/890582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/890582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/890582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/890582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/890582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/890582/1"}}},"commit":{"parents":[{"commit":"2897c21d8973ace33e29c38296362b4a3f7635e3","subject":"Merge \"Simplify test_purge_zones to fix race condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2897c21d8973ace33e29c38296362b4a3f7635e3"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2023-08-05 07:26:48.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2023-08-05 07:26:48.000000000","tz":480},"subject":"tox: Remove basepython","message":"tox: Remove basepython\n\nPython 2 is EOL. No environment should be defaulting to it. Our CI\nenvironments certainly aren\u0027t.\n\nChange-Id: Ibceb63090230aaf9348627e23ae55ca46764b9d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1fbd7f4843644c3dd4a548c5c19af4130be5528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1fbd7f4843644c3dd4a548c5c19af4130be5528"}]},"branch":"refs/heads/master"},"fd693fcb926d5520ac7d51e343e12ceff96fa40f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-22 21:41:30.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/82/890582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/890582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/890582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/890582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/890582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/890582/2"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2023-08-05 07:26:48.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2025-02-22 21:41:30.000000000","tz":0},"subject":"tox: Remove basepython","message":"tox: Remove basepython\n\nPython 2 is EOL. No environment should be defaulting to it. Our CI\nenvironments certainly aren\u0027t.\n\nChange-Id: Ibceb63090230aaf9348627e23ae55ca46764b9d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd693fcb926d5520ac7d51e343e12ceff96fa40f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fd693fcb926d5520ac7d51e343e12ceff96fa40f"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
