)]}'
{"id":"openstack%2Fdesignate~891926","triplet_id":"openstack%2Fdesignate~master~I30b967db85ab2c03f2cb159ad4df00365001c9ac","project":"openstack/designate","branch":"master","topic":"schema","hashtags":[],"change_id":"I30b967db85ab2c03f2cb159ad4df00365001c9ac","subject":"Remove unused schema validation","status":"ABANDONED","created":"2023-08-19 18:57:11.000000000","updated":"2023-08-19 20:26:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e526b3fafcc0bba323a76dc704843a369cf036f1","_number":891926,"virtual_id_number":891926,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-08-19 20:12:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-19 20:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"960833aa8bfa96ed40e8564202ff330c0b4c14bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-19 18:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e410882568757390a8fa14d167d0eae41acc9220","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 20:12:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fdf4e8058af4749bed7f4f6cb359261\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/68a50ad671a84dcaaf8fe8d2a7e61cb9 : SUCCESS in 32m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/08024d5ac8d04ac883cf278afc6a3cad : FAILURE in 15m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/155208c819664548bddde5bb854882e5 : FAILURE in 30m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7416045add8449f2a01fb1c155de29c0 : SUCCESS in 30m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3fb09de0d9774e578a7ae9872c11bb56 : SUCCESS in 44m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9aab76f6cba64017a7e8a7496fbcd250 : SUCCESS in 44m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fbdffa74f84247359ce21b61c3747df6 : SUCCESS in 1h 09m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/32328dc869314139b9b1f3db245983fc : SUCCESS in 31m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/96f74853f88440ac8d18b737ef297132 : SUCCESS in 33m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe699cd9c8a4d86b688dfb20260f04d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c681c999b94486796572799aa7abac1 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0044b9af5b54ae2908a9b38011daaf6 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6952f0ad221442b19047f847a75e0449 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2e551dc125a42788905e6373a3cfa90 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1384c564cea6459c91e0823996a08877 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf827ff109b4ff091bfd2ebba073e4a : SUCCESS in 8m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4dfb5a54a55e4478adbd7fdfa514785a : SUCCESS in 31m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8446889aecea40b0a3e6acf29547a910 : SUCCESS in 7m 12s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b93421c715974c4ea53fe5706c1bce57 : SUCCESS in 47m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/68e3e651767245d88343071232842256 : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e526b3fafcc0bba323a76dc704843a369cf036f1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-19 20:26:34.000000000","message":"Abandoned\n\nReplaced by https://review.opendev.org/c/openstack/designate/+/891927","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"863a4fc680817331c555c1606aabd6f8d449ac31","revisions":{"863a4fc680817331c555c1606aabd6f8d449ac31":{"kind":"REWORK","_number":1,"created":"2023-08-19 18:57:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/26/891926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/891926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/891926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/891926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/891926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/891926/1"}}},"commit":{"parents":[{"commit":"29097ed55953964557f28e75858bd211ce1c7bca","subject":"Additional recordset validation testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29097ed55953964557f28e75858bd211ce1c7bca"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 18:56:56.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 18:56:56.000000000","tz":120},"subject":"Remove unused schema validation","message":"Remove unused schema validation\n\nChange-Id: I30b967db85ab2c03f2cb159ad4df00365001c9ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/863a4fc680817331c555c1606aabd6f8d449ac31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/863a4fc680817331c555c1606aabd6f8d449ac31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
