)]}'
{"id":"openstack%2Fdesignate~891927","triplet_id":"openstack%2Fdesignate~master~Ia5e4a8760161424851eda22051385695ec2c7a9b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-22 10:38:24.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ia5e4a8760161424851eda22051385695ec2c7a9b","subject":"Simplify schema implementation","status":"MERGED","created":"2023-08-19 19:35:01.000000000","updated":"2023-08-22 11:59:06.000000000","submitted":"2023-08-22 11:57:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891927","meta_rev_id":"11ebc3644faf1e3808d0a90b0b917bda5aaf0fb8","_number":891927,"virtual_id_number":891927,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 11:57:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-21 23:50:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-08-22 10:38:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-19 20:57:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 23:50:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f671b49a30c5edca2abb39c9e549973557b0586f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-19 19:35:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"422a1e1ea038476d0b2fcc7fd5036e4fec5c5775","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-19 20:13:48.000000000","message":"Topic schema2 removed","accounts_in_message":[],"_revision_number":1},{"id":"dec551cad33ae3e0f1cb3b7fafd8daeab5fd4073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-19 20:57:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0db270d657f0404d946553ef28b9b454\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c9403eb396454a98a1ebea0f96b8d8e5 : SUCCESS in 28m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9a85f4a6a4374064aa11e1f12974b06c : FAILURE in 16m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/949314c492034433a1aa9105be1b6da4 : FAILURE in 14m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/584c6048d4234133bce263c10f17f872 : SUCCESS in 50m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8a5d9a927e2a4395903aca42afbe9783 : SUCCESS in 29m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a1a6f217b2744ef39eb2ef16f934925c : SUCCESS in 1h 14m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/809c37f9646b42b68b894db448a82f2b : SUCCESS in 47m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/61fa904a2e5541a6858e54d947899d63 : SUCCESS in 47m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b371d05ed56548d19ac208057ba55a12 : SUCCESS in 28m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb8a7abd90864934821bd81af2772b42 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1df43b0f1bb43969bcbe17a7b14831a : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a1faaec3bf440aaca6b9cf520f2928 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1046157ccca4814be2908e7865a2cbd : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18326775757842fa83425283d37f614c : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0b055d8d536489b8e59caa83d748e15 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8024a53a3698401c80bf8296995e8931 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b1a891b7f5b443cb98862c21fc888e2c : SUCCESS in 32m 19s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/790f0e8d626c439680a1057710a18dc6 : SUCCESS in 5m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8ca1997dbaea4b33a1c81df3124bd7e5 : SUCCESS in 29m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cc1c48237e1d446ca0c5ee4d21df79e4 : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50753e917f9141014721c56f90a38fa8ee5c7282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-20 12:56:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"57ea3df68f702be91523c40d3f41aaa7d6e33580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 14:15:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90714cf2ec784ae6b21a5dcb0cdb3188\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c086ca6632645c78b37adbc2f2ad597 : SUCCESS in 32m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/60f243f063474569bb96edf84f04b2b0 : FAILURE in 17m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a75af6449a4c4012af4e89ba338260b9 : FAILURE in 15m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fad832a9df1147f2a39e3f4d2d3e74c9 : SUCCESS in 31m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f01457ff9b234d2e8df3e98fb5c94b67 : SUCCESS in 30m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d49ded6b2beb48ecba351b997ddd5ee1 : SUCCESS in 46m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/718a916b5fbe4b0d812af70528485f58 : SUCCESS in 1h 13m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f6e0f83b1b7641e9950775085fcff61c : SUCCESS in 48m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/769b01ed53304c3a9036bc45012c4440 : SUCCESS in 36m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42c90eac4a484669b19622cea578302e : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c20443e500454f6a9b24fbb1147dbfac : FAILURE in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25b8d5523918490097afec4153958b42 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89401fcc62d24f438fc228ef13f1adf0 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7702a53554214a4f8bd019b2571556aa : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7d74b011dda4ad38a53517bdfe2299d : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df7191523ab4eff956d7c982cf6c9f0 : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7a5420d17c245e7bb11e07b58db5c45 : SUCCESS in 33m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d7e97ae20e0f4b60bc5507b445b685f4 : SUCCESS in 6m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a1a5b78d659d4ae5bb7d1aeb60b67e0c : SUCCESS in 58m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a24dee81a0414b218336392563ec902e : SUCCESS in 4m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4fa5da32ca2bc487c93262f0d39c53085a55cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-20 15:43:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5aa2530d27a1e6bace757785c9f5736c03c85ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 16:34:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f26cc3bc8e314baeaa33dc39d6eb9566\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f4f14928784f4cde825f7d0450ba9dba : SUCCESS in 31m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/94ae534a95474dcdb120be02565ebe86 : FAILURE in 16m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/da17d5db377e46a4aedd989e8af65a39 : FAILURE in 14m 54s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3ab41bda921746f6bda15f74fad8bf60 : SUCCESS in 29m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fb2df1b740944c7a8675ef987c61f0f3 : SUCCESS in 30m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9342243f3614446485f0ff188386cf4d : SUCCESS in 49m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1d460944f34e4bcead14e96e07918b35 : SUCCESS in 47m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/03415ea1ebdf45f788bc8163f2c51812 : SUCCESS in 29m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/67909380616d4387922cf5f8f6d08a68 : SUCCESS in 29m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c602acd8398f4f72814d49c248fce6ac : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d58f3a2c9c444d5fac8b75e042d134c6 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3e579796f7344f3bad552bef8642518 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8466bfeae35847cc8f663810c45edc44 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ae0cb9929b647459b6f06203ece004d : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a0189bd32c341a085bc273ada373cbd : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7166e9dd037342b49a01c687df2d0818 : SUCCESS in 6m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d9eb4a97ff444522a71ba412ccde1e53 : SUCCESS in 31m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2fca82a83c6e4e40929ff1e409417a7c : SUCCESS in 5m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c5d3accd80154741b2ce320ab74324c8 : SUCCESS in 30m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b5b0588f7a454ea2a5fb2a5c5d4178a5 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f79e742ac53c893ec943c2244b33361a7f05481a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-20 20:14:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8e4f13b0a1f77ded7a5a4570fa054e489cc363fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 21:12:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95f063fd0a274c0da7134586104e2cf4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9746f05928f34f8c912e48625abdf91d : SUCCESS in 48m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/34b0491897c6425ea3043d01f8ba054b : FAILURE in 34m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e214087f5f4447a9a873a8a7f2912902 : FAILURE in 29m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a73c6976230d4b9a945faa3ba0fea3c6 : SUCCESS in 29m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6ca52167baec442088bf2784a961ffbd : SUCCESS in 34m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9bf0ce4f6d1f4e728ccf74614ee2ad0d : SUCCESS in 45m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/10bd63f94fab4d9e935408dc62c04ca8 : SUCCESS in 46m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0d020208d3be4b61921af156970530fc : SUCCESS in 31m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a4133c7f224a4a8393f1bbf51d961745 : SUCCESS in 30m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca037917ab854f2c8d186ded0b863d88 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d7a3b6b72d4159ab0b5f521af31959 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7e0ec66d91442a6830f5a72ec060435 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bdc29b1fdbf4e598ad7b645bf0e1375 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03725d1d76e5483eb6447a51f51199fa : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e70532a6ffab4c76addfc5643308b4fd : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246bd563d15a4b66b6b85a59d15f0fae : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c6aaf9e8cf254fd585e467523fd2bb34 : SUCCESS in 33m 38s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7973ea19c8d749199ea204f88adebaa2 : SUCCESS in 5m 18s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b052eaf634164f59af3f0cd7bea48c85 : SUCCESS in 49m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ab74470aecda4e668006b6ad3df3044e : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"da18a60e09e1c8d84b65a1120b61ab6343a0fe4a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-21 23:50:38.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ca3efff9895dccb61117d9ba1ac846c47d759098","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-22 10:38:24.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"33fa622a92052cfd728093f837b6e30ab44a4e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 10:38:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cf8f511daa1063a39e903caa7b3091c0c49dbf54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:57:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e48f004cf6c348e99f2089a3510114de\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9dc3605dab9c497b9dd081c8973acefa : SUCCESS in 42m 48s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c713ce6fb4c48b08f827c7be51a67f7 : SUCCESS in 48m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd61c15b51a64df0ae5080c50cc5b92e : SUCCESS in 37m 45s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1965b50f6af64ab9afffa087a1ba7c60 : SUCCESS in 1h 13m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/680c0a178e0841bb91aff778fd409646 : SUCCESS in 47m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f58a62ca121646a6ac1f020e231fbfd4 : SUCCESS in 33m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6814f63d42f43f8b6ee096916637494 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24b82c545bf54c96b1f526730dfa96ab : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03fef652a50542dba506b8a5679f0fea : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f3a45a8eeb7426e8020b1e7ba312f12 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4848b738c5044d0eb3d0290dd6b7b710 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f492c893eb54c7686de31d3c266ead9 : SUCCESS in 40m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f966e9ce8d074e3596eb1333d20242eb : SUCCESS in 7m 18s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c5e35180101845edb52204d9f6af4e34 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":4},{"id":"ff70677de1fda8e0f181732c1c0c4c585ef16bab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:57:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"11ebc3644faf1e3808d0a90b0b917bda5aaf0fb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:59:06.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a70f1e5dbe94b9280345e0f6b306dfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf7b8c2fa704ec6945947b4414cdc56 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9d115c88a1d1e5e1429a4fa913209e4c7bb3531a","revisions":{"91a90327b063e835ac9d15245b971f7c6b1f5cf8":{"kind":"REWORK","_number":1,"created":"2023-08-19 19:35:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/891927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/891927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/891927/1"}}},"commit":{"parents":[{"commit":"18a5d4f4d9e6400fd1be97695b253d1acf920555","subject":"Moved all regexes to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a5d4f4d9e6400fd1be97695b253d1acf920555"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 19:34:55.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 19:34:55.000000000","tz":120},"subject":"Simplify schema implementation","message":"Simplify schema implementation\n\nChange-Id: Ia5e4a8760161424851eda22051385695ec2c7a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91a90327b063e835ac9d15245b971f7c6b1f5cf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/91a90327b063e835ac9d15245b971f7c6b1f5cf8"}]},"branch":"refs/heads/master"},"5fdab9f588c3474a17e9bfedfa9b2ab809879ca6":{"kind":"REWORK","_number":2,"created":"2023-08-20 12:56:47.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/891927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/891927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/891927/2"}}},"commit":{"parents":[{"commit":"18a5d4f4d9e6400fd1be97695b253d1acf920555","subject":"Moved all regexes to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a5d4f4d9e6400fd1be97695b253d1acf920555"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 19:34:55.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-20 12:56:32.000000000","tz":120},"subject":"Simplify schema implementation","message":"Simplify schema implementation\n\nChange-Id: Ia5e4a8760161424851eda22051385695ec2c7a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5fdab9f588c3474a17e9bfedfa9b2ab809879ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5fdab9f588c3474a17e9bfedfa9b2ab809879ca6"}]},"branch":"refs/heads/master"},"b2cf2b52f03000e6e98babe028a86d709df1d710":{"kind":"REWORK","_number":3,"created":"2023-08-20 15:43:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/891927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/891927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/891927/3"}}},"commit":{"parents":[{"commit":"18a5d4f4d9e6400fd1be97695b253d1acf920555","subject":"Moved all regexes to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a5d4f4d9e6400fd1be97695b253d1acf920555"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 19:34:55.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-20 15:43:36.000000000","tz":120},"subject":"Simplify schema implementation","message":"Simplify schema implementation\n\nChange-Id: Ia5e4a8760161424851eda22051385695ec2c7a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b2cf2b52f03000e6e98babe028a86d709df1d710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b2cf2b52f03000e6e98babe028a86d709df1d710"}]},"branch":"refs/heads/master"},"9d115c88a1d1e5e1429a4fa913209e4c7bb3531a":{"kind":"REWORK","_number":4,"created":"2023-08-20 20:14:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/27/891927/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/891927/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/891927/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/891927/4"}}},"commit":{"parents":[{"commit":"18a5d4f4d9e6400fd1be97695b253d1acf920555","subject":"Moved all regexes to constants","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/18a5d4f4d9e6400fd1be97695b253d1acf920555"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-19 19:34:55.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-20 20:13:28.000000000","tz":120},"subject":"Simplify schema implementation","message":"Simplify schema implementation\n\nRemoves all unused components of the schema code and\nsimplified the schema class.\n\nCloses-Bug: #2031922\nChange-Id: Ia5e4a8760161424851eda22051385695ec2c7a9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d115c88a1d1e5e1429a4fa913209e4c7bb3531a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d115c88a1d1e5e1429a4fa913209e4c7bb3531a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
