)]}'
{"id":"openstack%2Fdesignate~891938","triplet_id":"openstack%2Fdesignate~master~Ief4d42625bc100d3a881163c684f86e010237f82","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-24 16:27:45.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ief4d42625bc100d3a881163c684f86e010237f82","subject":"Added simple tldname regex test","status":"MERGED","created":"2023-08-22 13:19:49.000000000","updated":"2023-08-24 17:28:32.000000000","submitted":"2023-08-24 17:27:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891938","meta_rev_id":"4f7764bc6c1a3a220b257ae647490e35b9d529a5","_number":891938,"virtual_id_number":891938,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-24 17:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-08-24 16:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-24 16:27:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 14:49:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 16:12:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"877e3b5c975a9e34d1945393dc986e4b2e50694c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-22 13:19:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"804526acd8ee5254a489d408340390b8c873a971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-22 13:24:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"361a3be2a80d531318aa515ec222c1e41bb28922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:49:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e719ea23b9314d6c956f7f5193709e75\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d9cdbd96fc324e20a9fb63a8f4259749 : SUCCESS in 39m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/64b129ad70124f53962b1635c85591ea : FAILURE in 38m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a5754e0f6e8461ea7847b6989fc601b : FAILURE in 38m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7a02f5f70a914c2f96729cb8bc7566dd : SUCCESS in 35m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0d84373cb8114731babe66badbcb0d7c : SUCCESS in 55m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b2084fe60582426bbae0a0a7d5feb8d2 : SUCCESS in 1h 18m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a680806e62984d0cbe3ab3229453ed50 : SUCCESS in 1h 11m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f25bf0bda354f3a9c4cf8aac540121f : SUCCESS in 35m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5bfcc5543aa9443ba586134c02e75e20 : SUCCESS in 1h 01m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e273db2d750466eb139c2f500a2f5e5 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6941b42b342244e69b76d771cff5f9bd : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/933314ebefa24c12a14b5c6e7a4b2311 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2009d3bcbf6f496e8582bf6e13aebee8 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e363dca8c2547fca48a5f329a9b8a1a : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae4606c2a2794aafaeb1aade8b399b3e : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/105aacc42c224acb93c63bdd16c18f74 : SUCCESS in 6m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/44f9d2b101c04a37a226ac19da4e6d25 : SUCCESS in 36m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c851d29402ee42afa0082aca8ef5e97e : SUCCESS in 7m 17s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/44bace4d1c00490bb69081b1db9ee3f1 : SUCCESS in 35m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/399d951ab2164c399332509c979a05b1 : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"60d0c86e6a5c3be8c40f23a332c0d15a001b6049","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-24 16:12:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbb9b445c123efafa94100213f342ca5eeca83ff","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-24 16:27:45.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"857384b5a5fd1eb1dc99c620aab4710863c1e927","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 16:28:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"754b44c155d209df171e99520473b5658d72a33d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 17:27:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf483afd3354f8e84d0feaf1c47c6d8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fa2c63b1721c4691aac0534796d1c5de : SUCCESS in 40m 45s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8d518c20a46f42fc9108aa1678911efc : SUCCESS in 51m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/97e88604b2364914b79d74aabb36cab4 : SUCCESS in 45m 06s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0cb62afc79094b45a7114a2a1e9ae980 : SUCCESS in 57m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9942169fffe044d6bac7609977c47693 : SUCCESS in 46m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ceb809650aa34f8f8c8d22109d1ca3b5 : SUCCESS in 51m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f26bd6ecae54202b8949b0c7b739ef6 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e58d56ca6694b08be424f8cafb08f9a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6061f5eb62413daa9a91d4c8f2f9de : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7ac9e8c236848478dca16aedbb04886 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c52d10020843b1a9fdb69896b6dc59 : SUCCESS in 7m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc9dbdde35fc44e386bdfff655eae703 : SUCCESS in 37m 38s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/85463d7b073c4f8b88e93ec9c89eb751 : SUCCESS in 6m 07s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c94de6a8d3b94e918a17a0d88b66bb31 : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4cc498c6659f48298f93d429f6ecc9f8d6354ce8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 17:27:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4f7764bc6c1a3a220b257ae647490e35b9d529a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 17:28:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40f0543499444beb845997ff7d6ba151\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b975b489574247b4b5985067873e732f : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0cf0ccd9d83c543f42fac73c538d79b97b4d7c14","revisions":{"1f22c9862b69e8f0b56a8a75eac3100861b3e983":{"kind":"REWORK","_number":1,"created":"2023-08-22 13:19:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/891938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/891938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/891938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/891938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/891938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/891938/1"}}},"commit":{"parents":[{"commit":"9d115c88a1d1e5e1429a4fa913209e4c7bb3531a","subject":"Simplify schema implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d115c88a1d1e5e1429a4fa913209e4c7bb3531a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-22 13:19:12.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-22 13:19:25.000000000","tz":120},"subject":"Added simple tldname regex test","message":"Added simple tldname regex test\n\nThis test was removed during a previous\nrefactor.\n\nChange-Id: Ief4d42625bc100d3a881163c684f86e010237f82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f22c9862b69e8f0b56a8a75eac3100861b3e983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f22c9862b69e8f0b56a8a75eac3100861b3e983"}]},"branch":"refs/heads/master"},"0cf0ccd9d83c543f42fac73c538d79b97b4d7c14":{"kind":"REWORK","_number":2,"created":"2023-08-22 13:24:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/891938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/891938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/891938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/891938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/891938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/891938/2"}}},"commit":{"parents":[{"commit":"9d115c88a1d1e5e1429a4fa913209e4c7bb3531a","subject":"Simplify schema implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d115c88a1d1e5e1429a4fa913209e4c7bb3531a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-22 13:19:12.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-22 13:24:30.000000000","tz":120},"subject":"Added simple tldname regex test","message":"Added simple tldname regex test\n\nThis test was removed during a previous refactor.\n\nChange-Id: Ief4d42625bc100d3a881163c684f86e010237f82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0cf0ccd9d83c543f42fac73c538d79b97b4d7c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0cf0ccd9d83c543f42fac73c538d79b97b4d7c14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
