)]}'
{"id":"openstack%2Fdesignate~892543","triplet_id":"openstack%2Fdesignate~master~I8c721722eeb36de64b8c8a79d4fdcec0e92388fa","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-08-23 15:22:15.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I8c721722eeb36de64b8c8a79d4fdcec0e92388fa","subject":"Fix DB upgrade with using SQLAlchemy 2.x","status":"MERGED","created":"2023-08-23 14:36:50.000000000","updated":"2023-08-23 16:51:24.000000000","submitted":"2023-08-23 16:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892543","meta_rev_id":"ccb8c4c18ec1d51d0f32e039c03dcbec880e93cb","_number":892543,"virtual_id_number":892543,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 16:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2023-08-23 14:42:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2023-08-23 15:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-23 14:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2023-08-23 15:22:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 14:37:26.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-08-23 14:42:13.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-08-23 15:52:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8574af5b739939a1982bd20e3664c4a4f299639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 14:36:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"996cd766e6cc6ef60148b6a7b031d01adeb00fb1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-23 14:37:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8227090216b6ad89c9cdae5667cabac86197d8e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-23 14:42:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b2fa921a17fe41ebbbaef9a7c99cb87dd25b7fce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-23 15:22:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2af637012e04cc7c41b69659267607fc04417ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 15:52:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a98e08a56d2343f194d8a4405df9521f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3bfe58c42f7f43a58166f41ccc248e2a : SUCCESS in 50m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8eedadbadfd24d539b469fe91f70cc95 : FAILURE in 36m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e391c813f474daebd42edfc0dd4f395 : FAILURE in 34m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ffbd1a38486b4b67b74cf45131dad049 : SUCCESS in 48m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/edfa78db94f24911a622b0e68923b9fe : SUCCESS in 39m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/be1dec4447ab457dad049c96f708555e : SUCCESS in 1h 09m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eaa2b39af8814161a73d7f9682fda32c : SUCCESS in 54m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/18ad4a29321c457ebdc8fa6fbae18fd4 : SUCCESS in 47m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7bc3410f15e14f49893cc25030478159 : SUCCESS in 50m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50342740560d4ea687ccae84eb5aff48 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7814900722b649ea85820535706561b6 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32a75764fbc34b73a655d2560095fbd9 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c4a7f711a2e4f01a61ba1e417d31d45 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7bd1e373d5b43c48771cff2a47ec310 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a493e66618346eea213b879625f3b61 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28934bcc255d4f2982aa8f30d8d1b284 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eebff02a31624f2ba21cd0d17c76b205 : SUCCESS in 4m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4998b9c7bf824233abdda514e8c1dbf0 : SUCCESS in 42m 01s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/52e072ce8dfa4d239d3cb473d6949c91 : SUCCESS in 6m 22s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/34095cfa416d4a33a6ea3af20a6c929c : SUCCESS in 42m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fc6821131c6347e3aa4d87fc93add5d2 : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ac083d53777ccf4f044faf4af09f500f6e17691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 15:52:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"320e4e87bf42f7eb328f2ea39fb1930c20dd7b43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 16:50:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05c5cfbfdca940bf8a53134f52ade00b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dffe7fd0954a4d80a9e88e7309bc723a : SUCCESS in 32m 00s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8e28d3bbb77e49f183216fe2e40b7a79 : SUCCESS in 35m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c5c645dff82449e4be65b7eb24d5540c : SUCCESS in 42m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ae531b2e5bd454f8681eec7f0c434e2 : SUCCESS in 51m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7dcd15c9127484dae20de0925dd988c : SUCCESS in 44m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9caa3903ca474b868ce20e3dab3ec946 : SUCCESS in 52m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5983ea79b5564275b38de6a1456c4551 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b6bc6e987a449ba99f331eb77c4a31 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da71bbd64a2442d78f4b972cf53c2f1f : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d53e501dab92417dbbcdb5321f599058 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85631829c6504d8c95cc8d4bdbe1f697 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ead32f2296f546c5abfbad54d2151b30 : SUCCESS in 5m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9804edc6158245bb88dbe6d85fd0f89a : SUCCESS in 38m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/018c2ed39f634896b1bb052955636c70 : SUCCESS in 6m 21s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/91b160e1fa6a46bfa0443c1c650d14d4 : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"79e53f7eb42a5880bccb8f040597120513a214f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 16:50:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ccb8c4c18ec1d51d0f32e039c03dcbec880e93cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 16:51:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94c71654e7a24baea69785f84626f406\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5752a70eb14cb2adc0b7b7a5126870 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be5a3dd80c9048c3a2c41bcdc92eee56 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06482503881c89a709a1fe79c2c8d600244ad168","revisions":{"06482503881c89a709a1fe79c2c8d600244ad168":{"kind":"REWORK","_number":1,"created":"2023-08-23 14:36:50.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/43/892543/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/892543/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/892543/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/892543/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/892543/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/892543/1"}}},"commit":{"parents":[{"commit":"9d115c88a1d1e5e1429a4fa913209e4c7bb3531a","subject":"Simplify schema implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9d115c88a1d1e5e1429a4fa913209e4c7bb3531a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 14:31:46.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-23 14:36:48.000000000","tz":0},"subject":"Fix DB upgrade with using SQLAlchemy 2.x","message":"Fix DB upgrade with using SQLAlchemy 2.x\n\nSqlaclhemy metadata bind is deperecated from version 1.4 and will\nbe removed in SQLAlchemy 2.0 [1]. It can be easily replaced with\nmetadata reflect which works for SQLAlchemy 2.0.\n\nAlso current metadata.bind does not result in fetching all tables from\nthe database, so metadata.tables is an empty mapping, which leads\nto db_sync failures.\n\n[1] https://docs.sqlalchemy.org/en/14/core/metadata.html#sqlalchemy.schema.MetaData.bind\n[2] https://docs.sqlalchemy.org/en/14/core/reflection.html#reflecting-all-tables-at-once\n\nCo-Authored-By: Dmitriy Rabotyagov \u003cnoonedeadpunk@gmail.com\u003e\nChange-Id: I8c721722eeb36de64b8c8a79d4fdcec0e92388fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06482503881c89a709a1fe79c2c8d600244ad168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/06482503881c89a709a1fe79c2c8d600244ad168"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
