)]}'
{"id":"openstack%2Fdesignate~892609","triplet_id":"openstack%2Fdesignate~master~I0c7b70bde214da71e1bf3935f1e5f33107410c96","project":"openstack/designate","branch":"master","topic":"contrib","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-08-28 18:23:13.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I0c7b70bde214da71e1bf3935f1e5f33107410c96","subject":"Remove non-functional contrib code","status":"MERGED","created":"2023-08-27 11:12:02.000000000","updated":"2023-08-28 19:19:55.000000000","submitted":"2023-08-28 19:18:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892609-contrib","meta_rev_id":"7342ced313c734bacbee846f764065787affc923","_number":892609,"virtual_id_number":892609,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 19:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-08-28 15:19:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-28 18:23:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-27 12:16:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-28 15:19:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"351615be18b9ba254cb453f3bdc4f5818ea5c5b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-27 11:12:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97bafd4c3eae98ddc1787ae1ef66778eefd87509","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-27 12:16:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e737499a3c5d4ea2b131f2b47e981272\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9a444d605a374c4d923c9c0dad911b57 : SUCCESS in 45m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/691807435a9b42efa43d4d502b76b34e : SUCCESS in 59m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae442e6640f341949d725c5ee08c2356 : SUCCESS in 51m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e606844b70045e6a5740795f0d892f8 : SUCCESS in 47m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d647b28c83a340e08793a1f41585b4b8 : SUCCESS in 43m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7a02908bd6014a07865b9f041f402848 : SUCCESS in 44m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb2791846a3b48e1b5c80f4278710d7a : SUCCESS in 50m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c07170d11694bd6b97b86a270010375 : SUCCESS in 30m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/76495704f35544e1813a155a75f8fc2f : SUCCESS in 30m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f228930f1c7b429d926bdfd3a016143d : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe8456c6ee6145719022cc125d1a3fc0 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faaaaec49b7b4595aed2edb038577a9f : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df5de7909bd64b8ead7c37823b73fc6f : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c20e1104a474d8bb58ddc7314c8744a : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7a259765f43470bb89b3227cbbd7d86 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08d63d572e04404491bc8a41601f4b2e : SUCCESS in 6m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8f09b7dafe284d24bc0ed663425d0056 : SUCCESS in 2m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d592dfa31254a7ab5b62c4fce3f785b : SUCCESS in 34m 07s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d024aedcf32342b59d70fc087d0f6676 : SUCCESS in 6m 19s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/363b505b45ac4e178b19ba504456f3b6 : SUCCESS in 35m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/12dd71e1e0a6493aa3793c44193e1620 : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"865a23c56731e9d404aacaa6668cad30dcfdd491","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-28 15:19:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66c2d496bc3dfcbb1100bef2b5c2f2751889fec8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-08-28 18:23:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc404d8304a459f8c0a726591ebde4cf71d5715f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:23:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60f897e52829d7fdbe56aa8d4e63f0322195aeb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:18:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7aa7da10930440d9d973fea657f794b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/34efb7bcc3d047faae098a1bea41c472 : SUCCESS in 50m 11s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ae50e78d726b4445aba1400ae66e4519 : SUCCESS in 44m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/18dc5a0c30134ecaa5a29d78ff6381cd : SUCCESS in 49m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cdd3ccef25574dc581536a1b69fc495f : SUCCESS in 49m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e60d4ffe19b441b7a9b0b2075e5b6885 : SUCCESS in 55m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3f240cea38e84faaa20431676cccb087 : SUCCESS in 39m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17beffd721d4499398e5096a9e633a87 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd5a63e9560458caf3b36c6f5db31d3 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a1bf611071d47c5996ccfb10e4ebd89 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f096f1c57e94e04b563d96d217bef33 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fded8fd151545e7a5da5dc37dea2d12 : SUCCESS in 7m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f183115a09ad4fe1baad39aa91f1a080 : SUCCESS in 2m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bf82bdecb764469cb38fe071ad95aa9f : SUCCESS in 37m 28s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e2d8c4a9c1fd4b0b815d88cc8f33ec2a : SUCCESS in 6m 47s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5a3d5ca0656c4c5591fde7887cbfba72 : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ddedbf91d735895880d8b7130b7adb52cf56fac9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:18:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7342ced313c734bacbee846f764065787affc923","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 19:19:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa9993c0066041249bfd2ac0d7e897b4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20157b1444224519b4acd8b2110fe613 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"360433b38b449d1c53ab1357fdb0c4608c09efa5","revisions":{"360433b38b449d1c53ab1357fdb0c4608c09efa5":{"kind":"REWORK","_number":1,"created":"2023-08-27 11:12:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/892609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/892609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/892609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/892609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/892609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/892609/1"}}},"commit":{"parents":[{"commit":"4ed635f6207c7756b40e9969a227db882fd65e95","subject":"Merge \"Remove unused storage/quota attributes in producer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ed635f6207c7756b40e9969a227db882fd65e95"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-27 11:10:49.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-08-27 11:10:49.000000000","tz":-420},"subject":"Remove non-functional contrib code","message":"Remove non-functional contrib code\n\n- Remove code that no longer works.\n- Minor tweaks to existing contrib content.\n\nChange-Id: I0c7b70bde214da71e1bf3935f1e5f33107410c96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/360433b38b449d1c53ab1357fdb0c4608c09efa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/360433b38b449d1c53ab1357fdb0c4608c09efa5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
