)]}'
{"id":"openstack%2Fdesignate~892814","triplet_id":"openstack%2Fdesignate~master~Ib618d7850b0a86a8eb10eaa52b0e330cf908883a","project":"openstack/designate","branch":"master","topic":"catalog-zone","attention_set":{},"removed_from_attention_set":{"33634":{"account":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"last_update":"2024-02-27 16:44:15.000000000","reason":"Change was submitted"},"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2024-02-27 16:44:15.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-27 16:44:15.000000000","reason":"Change was submitted"},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-02-27 16:44:15.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-27 15:30:36.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-02-27 08:30:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ib618d7850b0a86a8eb10eaa52b0e330cf908883a","subject":"Add catalog zone support to devstack plugin","status":"MERGED","created":"2023-08-25 20:16:22.000000000","updated":"2024-02-27 16:47:27.000000000","submitted":"2024-02-27 16:44:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"892814-catalog-zone","meta_rev_id":"285d968c5a33b59d2d86c27ee428f4a6b3139013","_number":892814,"virtual_id_number":892814,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 16:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-27 15:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-02-27 08:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-27 15:30:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"reviewers":{"CC":[{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-25 21:40:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 06:59:23.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"state":"CC"},{"updated":"2023-08-29 08:02:59.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"state":"CC"},{"updated":"2023-11-09 16:31:28.000000000","updated_by":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"reviewer":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"state":"REVIEWER"},{"updated":"2024-01-22 14:40:18.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-01-25 12:40:49.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-01-25 12:40:49.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"e89dea63e26a0ecce358077ed49833c5ec9422db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-25 20:16:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a8c66ce96fc43dfa6ce6657cb6637faa0b13bd2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-25 20:16:38.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ad9a5ab2590546d424ffa47588b84d5a60cc11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 21:40:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e274cd42b024a7aa65def5bbc565a6d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b49633b6605a42078fad9a8712e6969b : FAILURE in 17m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8ca8c816cf3b44a79290a461ee84d8e9 : FAILURE in 24m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fac4df0a9a014915851fa1b4ff14d1ba : FAILURE in 21m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2fae5c6323694845aee2ace50050c49a : FAILURE in 26m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/233fd01e7f8f4836a10ad604d18243f9 : SUCCESS in 46m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/88ee1210b4794cf39b289015a09c177c : SUCCESS in 1h 16m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5a50e8b8e9034b9e84d55f20fd3e2cea : SUCCESS in 49m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d478e85fd64d4b8eb93a1e33970f7c2d : SUCCESS in 45m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e5618dc9e6f64fddbedf220bac3664da : FAILURE in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd658e136f97459bbd03ded1eb8803f5 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b46edcd801a456bbec8fb3fc39961b4 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f89a0a28bdb042e29d53e9f892eb6761 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8983fc7e1f948c1bab5a8162de30bed : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e55380fa306a4d21ae420c51ef5466f8 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df4e81efb20a47ce985eb1576e17392c : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340a06ef568645cd9a11956214d9c4c0 : SUCCESS in 8m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2e13ea0a024b4ee388f396876a831299 : FAILURE in 18m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a62e725605614b6a94cb55760301fb5d : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5ea0655db2654a8b9ad038571e970020 : FAILURE in 25m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6e72d5dd31fa47979c81a4bb234719ab : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f3fbbbf8004fd5f9108fe5d9e0eed4c3749f9768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-25 23:21:05.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"91ce362a3a7f71842aecf918621a9db04d19252b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-26 00:32:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d7c9d1bb97046ac8faddae02402aaaa\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e0a63f0389804f5990bad20598f291e2 : FAILURE in 27m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/17ced64778af4bdf8a9c8fc90b815cf8 : FAILURE in 35m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fff647578bc4f28992f4397327d2ce2 : FAILURE in 32m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/655707762c0b41bdbbc6ec7b88db6a08 : FAILURE in 29m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/81c79342511a4f35b57f46ed9b665b47 : SUCCESS in 36m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7101609e5e84485fbec68efaa1284071 : SUCCESS in 1h 05m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/03d34691a55f418f91c05bc9e5aad302 : SUCCESS in 1h 05m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/82f7f07604ac472cb86ea488adebc7ec : SUCCESS in 30m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/539ae57f8fc94acc8ff5734edc9b6615 : FAILURE in 25m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/233dc0e08e7049359262a9ec445681de : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f05066122c84174b3562261b95ecefe : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d501799947444f79ffcae8b27365dad : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/388fb1489fe54150bfb6128f13b4efd9 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d76d85e2ddc4aa6a5053a2f28b1b3bb : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35c13d283a6f4e0faf077b96f0e81e1b : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/144620e000cd4a348170443b91c1ce84 : SUCCESS in 8m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/505b9eb5606a477fbc23a45bc78e2e2f : FAILURE in 16m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/96a4ce6ee083481980e914a892807696 : SUCCESS in 5m 52s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2b80aaaec9fd45feaaa766b7f76b0e4f : FAILURE in 13m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0d60268316304736b3659d83b833f701 : SUCCESS in 4m 48s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d4ae44ecf65b464d9a584907db810b79 : FAILURE in 46m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0939a4208300082e9116ff8f2fb838c2eb4a498a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-08-28 15:22:46.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"fa113f144bc91ba30d249f2f69f13ba23f52af40","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-10-26 12:38:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ecdc268d0eff10afe787b439726e689326607989","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2023-10-26 12:39:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0b27ab0d6d35cb45f3351986c3bdeaea92429b39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-26 15:47:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5fce03b58cf604c8cec05b1ac4a210fcb781c56d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 17:09:53.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4ce0337c102e4ef8987604447f458470\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0ed5afb440e843399195b893b190698a : FAILURE in 15m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/17665254174d4881a7ab884bba009004 : FAILURE in 24m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a04cf77536a94ed284b003eb7582fc64 : FAILURE in 24m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b9525031ccab4cd9b4233bad51c6db50 : FAILURE in 15m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/30656d8e642447be87e08c8ef2f33092 : SUCCESS in 43m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3a01ce9f56a44a6fa1032b33f51b4112 : SUCCESS in 56m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/87cbb4f1b57e47a98e0fa24367341d6d : SUCCESS in 1h 13m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fefdf55c88e04160a06c8178e2f0c25a : SUCCESS in 49m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8a0027680636460a9cd673931a0bb4fd : FAILURE in 28m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/975fdcb01f594d41b5514ccb20c377db : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89372841a3a141a98e42d5ca57ba8f2d : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/544273776e6547c8a3b1b3da77d86054 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d66a62ea79354677913dd34ad3d6f290 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d169ea56c2d84a0387e34dff9f123614 : SUCCESS in 8m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e969517648104df89b5c90b34e654e50 : FAILURE in 16m 22s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/81afac67ed514faabec68601e3b7e9e0 : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/232405f1fc3649e5bd9c2815a24ab96d : FAILURE in 28m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9e9c23f20e834c0bb54dd91fae80fb0f : SUCCESS in 5m 20s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/699125c6c4904c10bfb1a232c654d81b : FAILURE in 1h 01m 23s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":3},{"id":"41d1b1b5f4e27151df31f0f0930e6f2f53946393","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-08 12:24:50.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a542a3b235ef1ac5c369d93ce62128290f1a1cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:47:58.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2f1bcebd8f74c338337bec9bde23fe2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84a756782362447ab1c8791c26923df2 : FAILURE in 18m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bee905b6815841938e0316652e906c6a : FAILURE in 22m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/364f865650804794ab9a1fcc171e9cf1 : FAILURE in 41m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/21acc04196ec40fda8df8bd0a2df13c9 : FAILURE in 19m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d256574a00064d8381fda866a062c21e : SUCCESS in 53m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9555ec6bc52d47bc91549ca7e68b7f65 : SUCCESS in 51m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/16e6c9d9de5c45cdb780a703501e6439 : SUCCESS in 1h 16m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ae488480456491cbb909d8191567482 : SUCCESS in 49m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a6af6dbdb1e64d869fb4ef131de2d254 : FAILURE in 14m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/682b909a7f7a48c1bae27fc6fc34eccb : FAILURE in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58d4eae91d7b42a2abf9ab276a6240c3 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe0d9fb04d348be98083febabcd4997 : FAILURE in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f738ad3fc4a4e18a96bd1b0bc2d0408 : FAILURE in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac278fe96c0748499da31f3874d13484 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d91fa60c7352461f91a4f58f458e88ea : FAILURE in 15m 44s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ed4e938264014bf18299a0169b339722 : FAILURE in 5m 52s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4685588c5d9c43a8a6fa2a5efd668283 : FAILURE in 30m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f72681ad327246f1a759b939814c0661 : FAILURE in 4m 40s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3314162df57a4584ae87bafa5602ad1b : FAILURE in 1h 09m 41s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":4},{"id":"8a7a34cdb8df702c24d33be3af2fd9dccd183fe1","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-09 16:31:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d2753c0ac5d1e1e09e82da07ea3a991cd8c83be2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-14 09:25:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"7f557242f41b155b2c07784804f9f94b3ba3a38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 10:32:47.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/600fa168791a4ef094490eba625d2f31\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/47c6a6e910e34ad6918e428ea381ede5 : FAILURE in 15m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9fbbbd51074a45ff85b6d47e3014ae26 : FAILURE in 40m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b94e09843c4e4643b858518de7635602 : FAILURE in 22m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f0537623c4a3416e9f3f9b735aad1f0b : FAILURE in 12m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e1764c03e4a9461f9af114451ac03aea : SUCCESS in 34m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/999f877902ed4d019efb64c10c439797 : SUCCESS in 47m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a065b22d3f14881891c639ee566181d : SUCCESS in 1h 06m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/de7c695755704f8cb92d0fef996b2260 : SUCCESS in 49m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a77028fe7b9d49d6b3a9d90ffc466dfa : FAILURE in 26m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d69c2a857536438598bd0bd3dd6fd9b7 : FAILURE in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8ea80a1afd4570add9e39c5837eff6 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63961cd58ff34677978a2b81b1c5f054 : FAILURE in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0363fba52d6d4639b34aae93a6a5865c : FAILURE in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1e9c384589c4edea6d6fa5c6f58ef06 : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7aeb828a1dbb4300a8a8b65c41ba4596 : FAILURE in 20m 02s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f19c0526c5404541ad5ba70fa08bfa75 : FAILURE in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b9a27bf9aac64fd5b21815ad4804dfb1 : FAILURE in 18m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9cb1f2e7b5334aba8cbd38edc172dc6d : FAILURE in 6m 04s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8962c3ad641040a9b0ede99bcf67acfe : FAILURE in 49m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"209b70ededd4ba532e34e1e622b09b68dbbf4c38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-14 10:36:53.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"796f69af4ed0fd6608eaf079677a71cf4ac6ab7d","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-14 10:53:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"faa0791831284abcb4278d126a0027597dc35dfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 11:56:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9875a61af15b4325aad4c6aed958896e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/97d4ba26c0fb41faaab07a47f47cc880 : FAILURE in 15m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/41841ab561ec41bca8cf1dc4e2c404a6 : FAILURE in 21m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/115aafd37f6d42a9aeba7c6b3d4e3962 : FAILURE in 31m 51s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2ad52459056e42539a85ac17139af458 : FAILURE in 29m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/311f365c4fbe4c1eb40f531966e479ce : SUCCESS in 51m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e15d7714adea4738ad2e14b287eca9ca : SUCCESS in 1h 11m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/29c818ae7cea43c7bfe04a74046a3501 : SUCCESS in 52m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/940239a569c04b2ca7746097709510b3 : SUCCESS in 51m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cd809ff9d8054ce884d8ee508acb3245 : FAILURE in 13m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42954aa53fdb4ae29f9563856f0a1de0 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6bc008565524f4c997dbe3742030d98 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c32f3c1fe5c8480bb674bc7c31dfa82a : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4a8aa32bda4472b820d7430d05fb334 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1b61b846aef409f96e45247f17e8ac6 : SUCCESS in 8m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/196289f664eb4e69bbdd319a782f6c36 : FAILURE in 17m 07s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/53fbc8cef4054f0b8c7aff63ef295b81 : SUCCESS in 6m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6542ce8f2a1d4c8680dd932172fcc14a : FAILURE in 30m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3de75072cc42494e9786ad1322f6c016 : SUCCESS in 5m 46s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e62fc2e9e83a490bb973dc2b6aa7296c : SUCCESS in 49m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cd7801c411a6435bf60d4eecad8754c07237ec00","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 17:47:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"949139b1b21071759b058a3c68598ce349410580","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-14 17:52:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3793f5a5358ccd0e3d94b364e9b04c920f1fcc61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-15 16:53:40.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4f0977b0763745f89511a3b04933c0ea2178b47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 18:14:37.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04613caa9a5649cb9753044e7120d3fc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/614a419fa7654f3c94fec6a2cd6f51e2 : FAILURE in 33m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1b72eab0f9254121ab2240ade01b6ea9 : FAILURE in 51m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d516b759a2d94d4d9d2a1a0b1c2981ec : FAILURE in 41m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9b478da6668542cf93e5d2bf04ce9024 : FAILURE in 18m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/02f8a03f92a74a67aa49a0fcb165b179 : SUCCESS in 55m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b0d7d3866dde4693a3df90c08af0932f : SUCCESS in 1h 11m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a148b7efc978466988e5bec0c91163d4 : SUCCESS in 1h 14m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/92067844a9d94686b07234956c7014c4 : SUCCESS in 54m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1406e66fb3bd400299ee144039a5e3de : FAILURE in 14m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efcb463c3d5c46928cf7207f9ddf0e18 : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/388afcf49ac54004a20618e1f6905861 : FAILURE in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84284ea77af94e0e8f49392142d1c98e : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2a49cd787984bd29fab90aaf3b33c8b : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7215d722fa4675b96a1def899f5afb : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ef5220caec145299917190de7dfe1ba : FAILURE in 16m 53s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6c914ac2f4a24afbba8455bc6258699f : SUCCESS in 6m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2bf88b3733f5476d874ccffcc084d93f : FAILURE in 19m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d6e2e8d90c5d48e8a044fa332300ca03 : SUCCESS in 5m 06s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7e8cd9d654ec4125b250710d84051999 : SUCCESS in 45m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6aeb086dd5034a4aa5826a701d6a72c3bcf0f3c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-16 13:02:50.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"806540fa789959253e565c33076057b6d21ac973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 14:30:22.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e39161c29024a058b63a3ef50c63425\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fc604d48407b473d8ae6cc363228a316 : FAILURE in 26m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d4a6b7880a42422eb9f622760f9a0644 : FAILURE in 25m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d508458a67a14f1ab0e85da364a914d5 : FAILURE in 24m 55s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/539c2304b6a34a17802c688b2116454a : FAILURE in 19m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/39b973a94ffd422b9d884e7475bbfb6a : SUCCESS in 49m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/320d7ea401a849b79304077e7ab94599 : SUCCESS in 1h 06m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0447bdd8f1ff4aa488f7e541bfa7709f : SUCCESS in 1h 19m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7be2f6d39fa24739b4217fa59e2156a5 : SUCCESS in 51m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bfe51bbd8df2426a8bf989fa2eda7313 : FAILURE in 13m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f9b16f46ab544fa85a27a2c82124ae8 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b976713528404aa9bfd99b565c382d7b : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61a816d8c77846db96b433f244f57cfb : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fb72065a36f64600bf01a9082954032a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f2dbfa273c4f2e9ae2c5fc8c657c42 : SUCCESS in 6m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78533272840940179da3328bad193688 : FAILURE in 21m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bd8dc7191a5c4d84a7305f71b650abbf : SUCCESS in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9e485d5d684a4211801b8ca5ad360817 : FAILURE in 28m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/04ab882188354b0da1771a77fcec9fca : SUCCESS in 5m 02s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/89dc72bcf0a143df9b3d84da1963b7e9 : SUCCESS in 47m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9e64028993260832648695e5d4bb290cdc874344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-20 15:27:52.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"b507d5174e8fb0736b59a8fb6c0ec9b35ba59b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 16:53:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ececaeea1edd498fb779636a4e7ca2af\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1f49d17a0b574eaaa9c238350a03b1b7 : FAILURE in 27m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b93bb11d813c4ed2a9b9ce7e3a2a2350 : FAILURE in 37m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c153c25424b4cd2b09b9d77206e7b2c : FAILURE in 34m 41s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/94cd3099eeba4d35805b5ff0450b3803 : FAILURE in 28m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/896311c09ed243b98b5c90d55cda9b39 : SUCCESS in 49m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4ca5c8f4b6e8488897220c3b9c996496 : SUCCESS in 1h 16m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/63ff904c5df14d1a86d2b71af81ded39 : SUCCESS in 54m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f4a45812cdb04a57acf5fb28d74f5bb6 : SUCCESS in 38m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/043fa5ab93f9437daaab4af992e473b0 : FAILURE in 14m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d42c09ad0284d35b6fe18bfd19d12ef : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6648dc636346b5987fdfb8b88b8d9b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d7a871127c41a8ba86bab4154322e9 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fafe81dd13d48c1aff32801468d6fe3 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/913eadd360e7412790b0648160e0a521 : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab3f7bc9588545bd89c5554f82754632 : FAILURE in 17m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/394c606aeb644f33af75573e406a060d : SUCCESS in 7m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/825cc441acbf46fbb1cf4abaa5b47ca4 : FAILURE in 16m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/72ee5837e11a41c6a48d408607ceb97a : SUCCESS in 5m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0fd185371fb14e629db9c0216d6dbcfc : SUCCESS in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"754152a252f7e0f1d39a6110e51a5989ccb472a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-11-30 13:15:41.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"be3ba98cba917285297712f5b222a391abc534f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:34:25.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb2883ab018444caad88376220b356ac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c5eee019dfce4d05a7df61a302f481ed : FAILURE in 19m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a96486b34f26402d896320fedd291495 : FAILURE in 22m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b6f8929f7c4461cb47f70b1a0e3142d : FAILURE in 23m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cae3d04675b44ee5ac1512bcefaca955 : FAILURE in 14m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/72737a897483498ab4d948afc2e9cd44 : SUCCESS in 41m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/61bfe2b69c73409faba893419a6655f4 : SUCCESS in 45m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e6589d18374f45be964b481a5bbfc46c : SUCCESS in 1h 09m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7bdd32d8b812481f84a09940e2e85942 : SUCCESS in 46m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/20d5c408df3d4ee48db9f0180038f447 : FAILURE in 18m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68f95b27bea548e79f6adb7cb6ac0728 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/024d36fbadc34d4b95453842bf166d62 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20f1feb9c41a4195b40960c021ac38fe : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76bb3d6a0d6a458895bb5ad4f4c87214 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6025925eae3b4c729207cedb13f259d2 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3cdeb1c6ab0f4f549bc178bea714a467 : FAILURE in 23m 20s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fed69463b7d6447e9b6526c0484562fe : SUCCESS in 6m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fa8c8994d3944432a49993a10f143415 : FAILURE in 29m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c555c99a5c8d4177958ce479b7d0821a : SUCCESS in 5m 27s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b92c4595ea9d43de93a94989ca99325a : FAILURE in 30m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5b296262b0979a4e56f8d1f69f0b4ee8ac257374","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-04 09:59:26.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"32b871073472adeb961f229987d310a338f7c87a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 11:13:41.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad67db1657b241a98f2124d8f5492df6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/71923c53341f4e2bb9dd2297ff969dbd : FAILURE in 19m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6196c4046e9f4282ae4fb2acac934243 : FAILURE in 25m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/35ee0b03a5e444caa3498d2ddb3eeb6f : FAILURE in 24m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/24e1ebd3663c4cebb446e901918c6ace : FAILURE in 31m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7a2d59262ad44130be15146a2cbec442 : SUCCESS in 52m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/84b2e9c445fa49ed8057b1b1aae2a753 : SUCCESS in 1h 07m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fa0f1dd8bca14ad5a101df22e250665d : SUCCESS in 49m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/280d45753a174e86a768713ee5a4b80b : SUCCESS in 58m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/798f04e3fab54948aa290f213d22d183 : FAILURE in 28m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fefdaa468ff423a80aa75dab09fd4b0 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f910fefd76484bafa59fd26d49335739 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b82eef56e583427481147439ba34f47d : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/066b16d087594e9093aa4e0e725ed754 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33dc319edef74f7c9972c4fa6658aad1 : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58427471f75b49768d7320c67f9ec329 : FAILURE in 17m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ac8c93a8af43427c998f2fe905811fc2 : SUCCESS in 6m 30s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/698a02ee542444b5bf583ecd2f9bb65f : FAILURE in 27m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d1e45a5be17641a58ab8d6c6bfe2bf4b : SUCCESS in 4m 58s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/285307842d2040a8b1932f41d6e411bd : FAILURE in 24m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d8d265ba4ac7c24cef4196a8ad8c70491ea21a78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-04 13:51:21.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"9c8f984dbddcbfb9fa53b3ce31f4bb8a0cf11c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-04 14:53:41.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9d3c3ae290d45e292abd87434b514eb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b184fe9be5d54dc49407e90adc1ab5ab : FAILURE in 18m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c18af4037e7e4575b86e1f4a5d315b44 : FAILURE in 41m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c98e37c0f8794e82970e2a09c0fd4152 : FAILURE in 33m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/64b46950598d43258047ca124b833c38 : FAILURE in 29m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/05744a97d1d549738b282b6b2eaf23f8 : SUCCESS in 50m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a14b627386344c8195de69598504f427 : SUCCESS in 52m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/48fd7edb71454b92ae9ad5eef8b5ed54 : SUCCESS in 1h 00m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11d614409f8e47afb14753e461b99d5e : SUCCESS in 47m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/117660fa15a74860b82cf6a1af939500 : FAILURE in 27m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e22a41b2a2c741f2b20c60d941f95206 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddbc54ce7734b09899e0a476b070cb7 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b864aa9486f41898b11400ce93e5e56 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7206724d9a654546a23cba13ced0ff6c : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9329406930c43ebb954c47b8788574d : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/49512253a68b4c8a940d4b807a36cc3d : FAILURE in 19m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d721e3b3f2f0476c9bb6339911f1ddd4 : SUCCESS in 7m 27s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8738dc6589314d5380b11a5eed711030 : FAILURE in 25m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/31e8be265cbf4d1da6d5cc4a596456ae : SUCCESS in 5m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5b2b080185ae4d5dbb6ac523ba2d7a43 : SUCCESS in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"2810ff39ddb304f79a9c2f5587821354e021419a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-06 13:19:45.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"203d6e85a26b9590178b27349f0df3ddd9cfc58f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-06 13:53:30.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"e193720a11729c40e2b6b2656592bcf613a54182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:20:40.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1586af3e1a99436296ea758794a1998b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cee31766eaba4a6e9ca721b7c4206b0a : FAILURE in 33m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0fe4364d35ee42308efa61fc42e831f5 : FAILURE in 40m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1fbaabd7fc24b3c8549d129d1219be7 : FAILURE in 40m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e1017d5654034bbdbfbbd960bb73b6eb : FAILURE in 31m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a79302758f9e4502bec35d0d51b2d6b0 : SUCCESS in 51m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/69ea7bdf37584cbd999f9927c867819d : SUCCESS in 1h 08m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68b927b1c19b40aca6ae9e8addd11600 : SUCCESS in 1h 18m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d08bbcd24b7493487f47ac618920a5f : SUCCESS in 39m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5bb2ab0a97534e01a8d560ac731ee927 : FAILURE in 30m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e729a56f1c64ff892202f6b85831091 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7d1aab462942119a5ab8d1d041261b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef516ef70d1249d0a86cd0048bea0fd5 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/184a4f57b85b46bca62c0692a3fde05d : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ad8ee4d51c649478ee574e066d614f1 : SUCCESS in 10m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ae5fdd74cb8b4152afe562df870fdd66 : FAILURE in 41m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3f23aa6a5f384d05a1b671afb50e9738 : SUCCESS in 7m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fd7d672530274f85b99753c5789b2724 : FAILURE in 23m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5f810b4ceedd45fcbeffcec899f25bb2 : SUCCESS in 4m 34s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/697916c7c22e45d0a648e5e73a6b8e41 : SUCCESS in 39m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6b6149c79782a845b155e84edd868af3a29b1f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-06 15:27:48.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"a9033cdc04f5ab469bc5bada610c407b71fc0602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 16:52:25.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb0a0d9ce5843c59683c914c02de1a7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d728f1af264a46fd8a38cee305022471 : SUCCESS in 42m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6abd16c6e8f241f18539ca75b46bcc02 : SUCCESS in 45m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a145531eb8414583844797060505a09d : SUCCESS in 55m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/457fca98c25546f78f2322d48ca20625 : SUCCESS in 36m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b8b0c1711ad14fdb83bd5d1be9f7e1b1 : SUCCESS in 50m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eb04c712c6984a2ba8db60e6f3ca9157 : SUCCESS in 1h 18m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/203e0e7e7d74490db0b007e51edec147 : SUCCESS in 1h 07m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d8632eb23c024e65bdfcb3e7c8230f06 : SUCCESS in 53m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5bb86348f774431cbb0ac23fb9d6c283 : SUCCESS in 38m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3ec4165fdbb40bc8bbd8f458ec1f351 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cac3e30460194f81b5fb292dabf28768 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02514c80f0044a7f8262be331d0606c9 : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d1f822ea661451385b66a1d74d6396e : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2397458b7c26489681893fd38a5cc103 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3098fc6447114adc8a007a0fb70e787a : SUCCESS in 31m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ed86de4267044e26937177f05e8eff2e : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d638c94113474166a2d589f8916554e3 : SUCCESS in 57m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0b552c892c484596a09fef57d3ae02f8 : SUCCESS in 5m 24s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6a247b70ad4d4e3cb7806631680346fb : SUCCESS in 39m 29s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2cf994e9be6bc73a1f572c2ef24ce0fdff0979ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2023-12-18 12:16:31.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"173b6106f973f997cb359bcfc00b551c5e3e8f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 13:24:16.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff077bcbca9404a9e85a18bfe4bf169\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cdd266e88d8741898b685eef962dd7f3 : SUCCESS in 51m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/30a3ca6bcdf44f788c10a84fddfea549 : SUCCESS in 45m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/000c9314c6bc4cee80ed16e7ebc191cc : SUCCESS in 54m 45s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4c97a9c0d93d41e18a22869c32d928f9 : SUCCESS in 47m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fc12de71ccab4a5087cbb390aa4aff7c : SUCCESS in 51m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7cf4c210a76042688424a9a702c9f0f6 : SUCCESS in 51m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f24c323e506a49ffa2d00ceea57f94a6 : SUCCESS in 1h 06m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/75049cb15fc44481a0ef043893eb7351 : SUCCESS in 45m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1cb850dbfdee4f5b9d3074d478c3608e : SUCCESS in 35m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a4f4fac9afb47f2ad0216586773b0b3 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89508e9d52c94919bd67a55d6b05671f : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55a5a9e2a1c14fda9567482e6c805702 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9e5ee14e60664454ab3e5e674e74d19a : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07fea1d3cc274ea7806f98e725ec58f5 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/42d0089094dc4b1fada75a6af13fb617 : SUCCESS in 31m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a71a98df03d7451fa6656a29d5882dc4 : SUCCESS in 7m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3baa7bc482424ea1af61b4b1f06051b1 : SUCCESS in 48m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a1ee9dca994e4b929699b51a3d8b6d49 : SUCCESS in 5m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c25030f75d19426cae93d515faabf06b : SUCCESS in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"915aa1a0b94f643fb78ba000a5f6c277bbb24c65","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2024-01-05 15:01:45.000000000","message":"Topic catalog-zone-devstack removed","accounts_in_message":[],"_revision_number":16},{"id":"f1a79b579d529d7d45c6b65de9746f6b7fd2bdd0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2024-01-05 15:04:38.000000000","message":"Topic set to catalog-zone","accounts_in_message":[],"_revision_number":16},{"id":"7fdc858df9663e9c9366e0ca0b6f65438878a937","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-09 10:53:00.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"7930f0fc1fa9637fa26c47518d6e25772c44b4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 12:13:05.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/689889e74baa44129d05fc39dc366154\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7e68bf04e9ff4bb88b2f928c7b6248cb : SUCCESS in 49m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/90d9f029a87343c0bc5e300b72854a49 : SUCCESS in 1h 04m 06s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e59093f0d0544724b696b8b981e59045 : SUCCESS in 54m 30s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9a6427b4dee14e1596c4cb5f119a7a55 : SUCCESS in 46m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0da9a91b1ee74b2c8e2757b1739a9677 : SUCCESS in 47m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e497697dddaa42b1a25a3113fba12692 : SUCCESS in 1h 11m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9750e5cc0bd643afb810531e5120589b : SUCCESS in 1h 08m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a8d2633f3d724c1580d5762b61d5d4c4 : SUCCESS in 46m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a0c554a4eb4444439feb05f2a7b1270c : SUCCESS in 44m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a32df5b2ef9342bbb7616313a711e3b0 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59b328a046624db987b77174578325b2 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/208d576ae4714417be99a26f3c0da1c4 : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fda770a7e2be4f7f8f4ba67f46ddb289 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75cc3e5752e4738bcc003301215d354 : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcc6161fb9b045c99691bd801e3ed662 : SUCCESS in 32m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d2a414f97a664203b45fdb44e53cdaa0 : SUCCESS in 8m 57s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2a5050b42b394d2f9f194c9b7fa50d65 : SUCCESS in 46m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/54a7187c342e4bbcb4dd55fb2a212576 : SUCCESS in 4m 48s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0177c5e0567f46648016f45e13cf4235 : FAILURE in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"bc4be3d32e7ba8431e9acf69742ac5d491c104ad","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-09 12:36:09.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"387b61301b5f9a975d0ffa3cfc399d1616115c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 14:06:56.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442f603ebf3f418b874b3f71e25d2d27\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2819294617ac4756a4c9c4030349544d : SUCCESS in 39m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f877a5eca2fe4c22a6decf7ec1334a02 : SUCCESS in 43m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7771eb816b6c4780af5e0ec86372a13d : SUCCESS in 43m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ae5e5f3191114c95be0c6cb9173a4406 : SUCCESS in 36m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7802ba9ab9a84c9083b6dc027253ba28 : SUCCESS in 41m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba0640a8fceb4659ac296fe5e34ca913 : SUCCESS in 56m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/525657ca902949a581ca6f0a7e207735 : SUCCESS in 1h 14m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/293c6c5177ce4acea8e6b78a16bf74f4 : SUCCESS in 48m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a5124f8ce79f498495bd75ad7ff955a3 : SUCCESS in 50m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64439b0979254af2b5ebf586de8f2033 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6f89ccf33f44008a9c0e51dc285f5cc : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b084c6b9cae4f87a5d56da2b184825c : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e24ec440b2e1446cae62f9f077914fd9 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/974aff6df6414ae58ea67b8d376eab4f : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a29570e2b50648b29a7ff4469ff5b9da : SUCCESS in 32m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ff45bb3f79ed4e3aa314fad0675af1a0 : SUCCESS in 7m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d02f1d12f5ed49409d92144134ccb94e : SUCCESS in 45m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e4d18b92dbb9451eaf9de45c3ed3d589 : SUCCESS in 6m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7a06137c3df448c8bbbf8fb4dc07e953 : SUCCESS in 48m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8fd66b002f8ddc4ebb30feac4089867dbd1841e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-16 14:17:52.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"69628e0da56957a63187ee0b4cf8a33a6c8a7fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:27:50.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10dd44e686a040c18e9463fd1b3745ba\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f26770fb10cd4f3f88255c3a450ad116 : SUCCESS in 40m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/93e2497e98f343b9bd66c58227fd4248 : SUCCESS in 51m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2420b9466b94508bbccb1ddd62741d4 : SUCCESS in 43m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3442bd78a674435aa3c39b442ecfecf8 : SUCCESS in 40m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e498cdd078df498cad063f732fe44884 : SUCCESS in 35m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fa3467b3e2004d95b6a95c59827cc0ac : SUCCESS in 1h 01m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/726f3bc74ac046bab806836b2723f27f : SUCCESS in 1h 04m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/90eb0d9852a9437d96e52b1e60143e3b : SUCCESS in 43m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d02b8b8d06c44a9387b7197bdc9daacc : SUCCESS in 45m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66bb53e59f934d26b44247c4972c3e92 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e325e1b7494d48bfb498efac350ca7 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0853c72c663c41c3a6fb1385be4f7670 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb20410c0f3c41438f41b54ea9c58d76 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cba2606cf7499095d5871576827caa : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0e57ed0a25a64f8f94e15d67aa60f603 : SUCCESS in 39m 32s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c03695b3ad334ceb8a745431a1cab012 : SUCCESS in 7m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d4075ee193874fbfa6617a3675b4fbe3 : SUCCESS in 1h 00m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/49f9801f0d704ebd843a37cd169e0d5f : SUCCESS in 4m 53s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0c1e17ac558b42f2a171761d0ec1c7e2 : SUCCESS in 52m 33s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"bfcb2d1c74ef5f91d680be9570b0f09205394131","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"date":"2024-01-22 14:25:27.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"1f3e56ad19e198ee2b56fea457d424a50a1a9857","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-01-22 14:40:18.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"d823c9ba0dd55140462d617c248e0a4b74f160d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 15:57:52.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b113920a5dbe4fffbfac83fd606c09a8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2131e5b471a84c8babf1ce2ca01d9823 : SUCCESS in 50m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2bd031b784ac4850838e82b414d2dedc : SUCCESS in 1h 02m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7e8f634c5d140c0a11a130873590c94 : SUCCESS in 58m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9cdad19247e74c0c855397f37ff5aa1e : SUCCESS in 52m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e0dd28275309442a965c7aa69d3694ed : SUCCESS in 53m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/06f2fae1f0ca4378883ac84b7fb5f99e : SUCCESS in 1h 21m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/76f8266ad44e49348109b5c276b18493 : SUCCESS in 1h 03m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/13d798d00b8c43baa0ccb4ba4ab97c85 : SUCCESS in 49m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8849993ae46b43c9be6c8f4bf8ef15eb : SUCCESS in 50m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84e06b364d0747df8cd9628151e76a23 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db027b54b190476e95d8fea55d33e9b2 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/160c53f0a59a464381785be32e1730e2 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28a6b5d4e27e48b5979682096e8d6ec0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98cfed0d0ff45e185968ddc4ec2387a : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d7a61d3e1324c24b0e72513aef82bb1 : SUCCESS in 32m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e7bd042e562340ae930eb04ed6b87b29 : SUCCESS in 8m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/904555b7a9b94d1ea908b13de29e5c05 : SUCCESS in 40m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7b75d33c37424e44a0552ee815647473 : SUCCESS in 4m 55s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9e60733a51b14c679d69cd8ec26d07fb : SUCCESS in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"022e8b7047c9b6f4a833f6920a5fb937be70c956","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-01-31 09:43:45.000000000","message":"Patch Set 19:\n\n(5 comments)","accounts_in_message":[],"_revision_number":19},{"id":"84d855a8e6608415b8122a920ed98a96bfb7bba9","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2024-01-31 09:53:55.000000000","message":"Patch Set 19:\n\n(2 comments)","accounts_in_message":[],"_revision_number":19},{"id":"c9c9dd35c9b16f0a479ff07d11ce8378a7bde648","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-14 18:58:20.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"07e26980a95a9bc12b1762bb3963ed22a65b2b0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-19 16:40:19.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"52d1e735c93f7f6af2ef9f4f39ebfdfd9d1ec1f9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-19 16:40:22.000000000","message":"Patch Set 19:\n\n(3 comments)","accounts_in_message":[],"_revision_number":19},{"id":"7e231e420004adc7c8e94b91f3ef7018d3cb7dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 17:42:39.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa5cf338392a4b43852b1d8c1d7c65b4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/08756c5c6f564072b64b5b1015207f61 : FAILURE in 7m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13d9936a58314e8687e7cd3bab2d41e8 : FAILURE in 8m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c56cb308d56640ce8ed0c98258ea7365 : FAILURE in 22m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/451bb160bbb14eee903f080b667f86cc : FAILURE in 18m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f98317930694b74910c973ba9f4465f : FAILURE in 14m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8bd6e5b2748648b29463d78005e3d9e2 : FAILURE in 48m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9b41e086df024539a7235e9419a0b937 : FAILURE in 52m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/44423f4e86cf405bba6a79a6cf1d02bd : FAILURE in 5m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c816b607e551425597ac3457104d20ab : FAILURE in 16m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d5c7a8bc74640a48b29199d716d699b : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6226031cbf4fe98f68dc18870e6317 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165a4e4d05b848e4b3beabb367fe89ec : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f51677affc9d41caa419a841cc206f88 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217fef51b1f04bf086336ed8b7f20093 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0eb3c53b853b4f7582e07a34403561ae : FAILURE in 7m 35s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6342986d3c3e441781d7649f65357efb : SUCCESS in 7m 12s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f0f04dc06a3b469fa15ec4bb22f8a3d7 : FAILURE in 6m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bdcac13c903c439389f342e147aeb697 : SUCCESS in 5m 15s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/98af1a02091e46df907280850ce34264 : FAILURE in 15m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"1e4ae6c600d99bca2d3c2b938b16212ebc96c29b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-20 15:10:20.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"a71f31c340be50bbe7fb9e27dd73905596be8290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 16:07:26.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/217d3e9ae2e140f98de01ac54967989b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9c9eb4370e42474694ce17f1314f3853 : FAILURE in 7m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/84b6f252390246548015303ca902c3c5 : FAILURE in 11m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b035d95312b4815ab2912468fc0b21f : FAILURE in 9m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b33d41525e494263bffbf9c9d73af1ac : FAILURE in 17m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3407089b0ea04d409c6276079433631e : FAILURE in 5m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3fb408110ad54f08bf18f6e78241e4d5 : FAILURE in 42m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d1e732c89d5041699d8747682f21c147 : FAILURE in 51m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4af2fd07db82450983b5c6ea73ff7bb1 : FAILURE in 16m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0405014f14d541308d64957f943fa8f8 : FAILURE in 5m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2fcabd5df3d4754898150ba9ba0ea7b : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2260efd43a94482994f11c483c1ec39 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa9506980e1b40d5a449e07eb662f7a1 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc3e1055a769405aab8f6c608a14233f : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff26826acfe47189d6c56aa87fc3711 : SUCCESS in 7m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efb011b6b6c148729a894bb115ad0d8d : FAILURE in 7m 24s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7ff2f6f5d15247aabe07fbba98206330 : SUCCESS in 6m 35s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d99b3ae21050495c97cadf7985d68fee : FAILURE in 16m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/20464f5b50774dadb4afab02e4568eeb : SUCCESS in 5m 03s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/626dd7113c78427ebee40d7852c49c11 : FAILURE in 16m 50s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"25882621d187243290c8b6b3aa29e2d5b0bf220e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-20 18:24:43.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"6d137767078f424f1d7fafddd7bcd80261f358e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 19:49:17.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd41dbe27db545fbba30d5bfc5d9a3a2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/979726a346fe4dc9b40af6232546b1f4 : SUCCESS in 39m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e8915480390044af96303d785d535a4f : SUCCESS in 57m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/eec439219cd2441c845733e4fa2ea3bf : SUCCESS in 38m 14s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/751fb0172f9f47e0bb10d1cc37c7f9ae : SUCCESS in 38m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/42cf8cb39374453cb79675fe7dff3236 : SUCCESS in 55m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0bc767c299f54274924a09cb71a115b3 : SUCCESS in 1h 09m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5aed3f168a840b6838e48f51b30fc0a : SUCCESS in 1h 12m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5d4c2c9104d74e4cb1aa1f8f874a6be2 : SUCCESS in 31m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4066cf6508c74f6e9a49b139d88489f7 : SUCCESS in 33m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1656413317204631b3a01e6b954daeec : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76668ec1b1640a5a59a3fb7bca325c1 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/045655488fa04d759b9ee7e93dc7e855 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1015ca4f24544e8abe0037f2a10a315c : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4eedb6e5830b4f7bafcf9ce94ab73220 : SUCCESS in 8m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c459940bd8854098a98a93d24afd8334 : SUCCESS in 32m 32s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be8e9ecacfc04be6921ecfa3fee0a11e : SUCCESS in 6m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c91d678e07ff4dadaf36693d0b63c024 : SUCCESS in 37m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a896b5829b674aeda34265cb76ac595a : SUCCESS in 4m 02s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8e44b5d71c6e46cf8a88a172d25b739d : FAILURE in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"ba835f14fec6af249b9b31e0d18881238640ddec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-20 21:00:11.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4017dc94f6ca5757293b4f9322c8ef6bf9ee1bc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:16:55.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84faec0cedbb459db56b120c5baaa014\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ced282eb400a492ba99a051ef55a4a34 : SUCCESS in 40m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b5cfddf2c7349699d49fe6c0b1bed1f : SUCCESS in 43m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/53db7f8bed614f0b914b419cfc5b81fb : SUCCESS in 49m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/86c371d3d18c45d281671a351fc68452 : SUCCESS in 46m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dc417f3daafe4f249a67339214cb7b8c : SUCCESS in 52m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6c6746f71dc647f1b6b53f96e64a4598 : SUCCESS in 1h 10m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4917480a70584b2bb90af06e9eae99b9 : FAILURE in 41m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c72d849cedac41c29e954e9321927857 : SUCCESS in 47m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab60252269db4a379b52b2d76d0a38e9 : SUCCESS in 46m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17a03ae5a2d34682a1387060f9ee0baf : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26a4180c369247b3bdff3c018a086a8e : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa533805abbd42dd9cd756a0ec1f7bb2 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a6bc25ab07f43918a445f4a8b9d5637 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c1361655c54ef6803996f639dc90a0 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/424e130927d144779d3cbaad2eeb8c8d : SUCCESS in 31m 56s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3be97781cd5e4df98a7f1f920bf29e8e : SUCCESS in 6m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/927fd7518a1d43e591915fdeaa3100ab : SUCCESS in 40m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/847a7b988ef94e0693754aba40a14e50 : SUCCESS in 4m 32s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/baa458381ef448b594690ada7c92a75c : SUCCESS in 48m 31s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"d0010417ee44dd141c7fdb0794517e00395f2fc4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-20 22:24:42.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"9fd388314c9d72af043390518ebc8cc7a720fc74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 23:45:38.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0907c5dd097b484a96c861ab67398b56\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fc26753867244658a8a3c55370004cbe : SUCCESS in 35m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e70e1e9c47a149f699002614d96655aa : SUCCESS in 43m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/24c1ce398b9c47d2bed1ebd7dc7f2978 : SUCCESS in 52m 19s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/af5ea2ae15264eb78188dd0a8487820c : SUCCESS in 43m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/35dcb58a71a94327bdc774eedcf248e6 : SUCCESS in 54m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/11f7f074fc1f46dc98f57b091b0f9c76 : SUCCESS in 1h 14m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c148e11467645eda4a8284c3c95f23a : SUCCESS in 1h 06m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b033f905dae04046b21ee4e540e1dea9 : SUCCESS in 37m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf567b732fc047088b103ccbfc9bf799 : SUCCESS in 32m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/255afd146c18403db61441e9fd5c90c0 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e302aad5c34c21a495a5349ba3d462 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12c42896eef34ef8953dc50590b57024 : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/465b4e4748494002ae9ccf27911a56bb : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c8c557ab244b48af768a1162dac163 : SUCCESS in 7m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bca63e16d13644469e6a0fc020a643ac : SUCCESS in 36m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/946ba875edaa454eba239e7c053db1b6 : SUCCESS in 7m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/34bc6f6cdb924c469b3e7ada3411ea7b : SUCCESS in 35m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e2f9e697270544118fda015febdd8e93 : SUCCESS in 5m 47s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/134ca57fc8ee45a3a8cbd91448fd63c1 : SUCCESS in 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5a8763d91876465f417f1849b76617f402992177","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-02-27 08:30:11.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"c93145c55cbd26a630a9789e8c4659570f1b129c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-27 15:30:36.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"54333b76503f40e67a721ed1dd8156de043a0c42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:31:19.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"c931d178c09c7b3596715ddc1b290b2f5688b2db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:44:15.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4cc1edbc59467483292e8e1920335d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/79c599493973437188357fbff92f93ea : SUCCESS in 42m 40s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/45c571628fe04a1388b6e3c89078d01d : SUCCESS in 38m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f90ccce387d64a67adaf248fa7b972de : SUCCESS in 38m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5811f3f61b348ba8349081396515c82 : SUCCESS in 1h 11m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/67f4f2e621c041c5ab0b2ed31a3ea9fc : SUCCESS in 39m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/12f5541fd9864d569d3dc22346244e22 : SUCCESS in 40m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f511e355ae4e288fc3585e640c7809 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/541e610a592f4070b490a7ab9569bb77 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42d163b532d742c4a4095c1b32fb1075 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9462e05701cc495b8f1bd1d652ec41e4 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7dc4226db2bd46e4bad2a39c0f9af8ec : SUCCESS in 44m 02s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d29e91883f8b4de38b9b158a79518279 : SUCCESS in 6m 42s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/290ee18f00d64e81be02a9706673ae77 : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":22},{"id":"5c4484319f1ee5250a6b27c722bbdf773e96f006","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:44:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"285d968c5a33b59d2d86c27ee428f4a6b3139013","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:47:27.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9e8ffe0d86c4b1daf2bafd77cc3c9a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c08e97d91f20412dae508e709ed59048 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"4258633ae5713b2dbec3bfe1a047bc76fd7a10f0","revisions":{"e85dc103f4a0319b6129dc585c3e50e5258bf73e":{"kind":"REWORK","_number":1,"created":"2023-08-25 20:16:22.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/14/892814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/1"}}},"commit":{"parents":[{"commit":"394d09c3c533c4d765a0a01b259336c31a0cbaf7","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/394d09c3c533c4d765a0a01b259336c31a0cbaf7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:16:15.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e85dc103f4a0319b6129dc585c3e50e5258bf73e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e85dc103f4a0319b6129dc585c3e50e5258bf73e"}]},"branch":"refs/heads/master"},"9a1811170e2bc3d7529a35259217a90ea9258f0a":{"kind":"REWORK","_number":2,"created":"2023-08-25 23:21:05.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/14/892814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/2"}}},"commit":{"parents":[{"commit":"394d09c3c533c4d765a0a01b259336c31a0cbaf7","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/394d09c3c533c4d765a0a01b259336c31a0cbaf7"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 23:20:45.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a1811170e2bc3d7529a35259217a90ea9258f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a1811170e2bc3d7529a35259217a90ea9258f0a"}]},"branch":"refs/heads/master"},"1e7a9b3c1ce035ab131dffb9dbe62b83cdf2851e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-26 15:47:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/14/892814/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/3"}}},"commit":{"parents":[{"commit":"73ef8976ef50661533871be1aa6b37468e896caa","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/73ef8976ef50661533871be1aa6b37468e896caa"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-26 15:47:56.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e7a9b3c1ce035ab131dffb9dbe62b83cdf2851e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e7a9b3c1ce035ab131dffb9dbe62b83cdf2851e"}]},"branch":"refs/heads/master","description":"Rebase"},"65a40bb43ba673c79772bf4cb3e5768b23d5b46d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-08 12:24:50.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/4"}}},"commit":{"parents":[{"commit":"af3a3db700fc787a9dcfca268e9f0633784ee1fc","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af3a3db700fc787a9dcfca268e9f0633784ee1fc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-08 12:24:50.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65a40bb43ba673c79772bf4cb3e5768b23d5b46d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65a40bb43ba673c79772bf4cb3e5768b23d5b46d"}]},"branch":"refs/heads/master","description":"Rebase"},"a0eccd798e6e27edcb850f330e41a5ced0fced70":{"kind":"REWORK","_number":5,"created":"2023-11-14 09:25:14.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/5"}}},"commit":{"parents":[{"commit":"af3a3db700fc787a9dcfca268e9f0633784ee1fc","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af3a3db700fc787a9dcfca268e9f0633784ee1fc"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 09:18:56.000000000","tz":60},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0eccd798e6e27edcb850f330e41a5ced0fced70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0eccd798e6e27edcb850f330e41a5ced0fced70"}]},"branch":"refs/heads/master"},"1e3d59639513dcda8f083c26cc2a8a81208c432f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2023-11-14 10:36:53.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/6"}}},"commit":{"parents":[{"commit":"9717f1503956ef422da758493019de5557ec5657","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9717f1503956ef422da758493019de5557ec5657"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-14 10:35:23.000000000","tz":60},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e3d59639513dcda8f083c26cc2a8a81208c432f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e3d59639513dcda8f083c26cc2a8a81208c432f"}]},"branch":"refs/heads/master"},"784b4299986b12fc485f12baa99161359b8532a2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-15 16:53:40.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/7"}}},"commit":{"parents":[{"commit":"ad8058b8ff1c2f9eec867d3c0f233857a6a23492","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad8058b8ff1c2f9eec867d3c0f233857a6a23492"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-15 16:53:40.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/784b4299986b12fc485f12baa99161359b8532a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/784b4299986b12fc485f12baa99161359b8532a2"}]},"branch":"refs/heads/master","description":"Rebase"},"5039ca28d7164ca5fb841cd86f998bd942098f12":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-11-16 13:02:50.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/8"}}},"commit":{"parents":[{"commit":"8a1117b204c2327c93cfda725a8f6110e32c79ab","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a1117b204c2327c93cfda725a8f6110e32c79ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-16 13:02:50.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5039ca28d7164ca5fb841cd86f998bd942098f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5039ca28d7164ca5fb841cd86f998bd942098f12"}]},"branch":"refs/heads/master","description":"Rebase"},"57a548d6b4b2495d23bb222f09fb0694fd099a3b":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-11-20 15:27:52.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/9"}}},"commit":{"parents":[{"commit":"116fcde3bf011bc2bce02cbced13994f1bc0d6a6","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/116fcde3bf011bc2bce02cbced13994f1bc0d6a6"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-20 15:27:52.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57a548d6b4b2495d23bb222f09fb0694fd099a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57a548d6b4b2495d23bb222f09fb0694fd099a3b"}]},"branch":"refs/heads/master","description":"Rebase"},"d3a25200cd26bb0d11d33603fde8109d79871d5b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-11-30 13:15:41.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/10"}}},"commit":{"parents":[{"commit":"5ac83ea10b40a98eb423cf124d254204d04536ef","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ac83ea10b40a98eb423cf124d254204d04536ef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-11-30 13:15:41.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3a25200cd26bb0d11d33603fde8109d79871d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d3a25200cd26bb0d11d33603fde8109d79871d5b"}]},"branch":"refs/heads/master","description":"Rebase"},"b50a96684d2cce3f59b4bb203d399ea03677bf0b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-12-04 09:59:26.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/11"}}},"commit":{"parents":[{"commit":"8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-04 09:59:26.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b50a96684d2cce3f59b4bb203d399ea03677bf0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b50a96684d2cce3f59b4bb203d399ea03677bf0b"}]},"branch":"refs/heads/master","description":"Rebase"},"6e1f47eb85a9cd084da7748ea7f3a67d49ddee84":{"kind":"REWORK","_number":12,"created":"2023-12-04 13:51:21.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/12"}}},"commit":{"parents":[{"commit":"8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-04 13:50:05.000000000","tz":60},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e1f47eb85a9cd084da7748ea7f3a67d49ddee84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e1f47eb85a9cd084da7748ea7f3a67d49ddee84"}]},"branch":"refs/heads/master"},"20dd433dd86ea60f55a7234913d97b0f945936b9":{"kind":"REWORK","_number":13,"created":"2023-12-06 13:19:45.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/13"}}},"commit":{"parents":[{"commit":"8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-06 13:19:26.000000000","tz":60},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20dd433dd86ea60f55a7234913d97b0f945936b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/20dd433dd86ea60f55a7234913d97b0f945936b9"}]},"branch":"refs/heads/master"},"16fd357ad67c96fdd97237b124c61302a0bf45dc":{"kind":"REWORK","_number":14,"created":"2023-12-06 13:53:30.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/14"}}},"commit":{"parents":[{"commit":"8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-06 13:53:21.000000000","tz":60},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16fd357ad67c96fdd97237b124c61302a0bf45dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16fd357ad67c96fdd97237b124c61302a0bf45dc"}]},"branch":"refs/heads/master"},"abc39b10a6f3439447d59bb44c0dbcd2af6e70f1":{"kind":"REWORK","_number":15,"created":"2023-12-06 15:27:48.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/15"}}},"commit":{"parents":[{"commit":"8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e76f2f6caae51dc6a38d4ba5f4ecc70e0b204ab"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-06 15:27:39.000000000","tz":60},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/abc39b10a6f3439447d59bb44c0dbcd2af6e70f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/abc39b10a6f3439447d59bb44c0dbcd2af6e70f1"}]},"branch":"refs/heads/master"},"ef07b0b2576451a279e2875800e565616822a03f":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-12-18 12:16:31.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/16"}}},"commit":{"parents":[{"commit":"81e879526528f96c5efd1a0c76dc628cc2ed7880","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81e879526528f96c5efd1a0c76dc628cc2ed7880"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2023-12-18 12:16:31.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef07b0b2576451a279e2875800e565616822a03f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ef07b0b2576451a279e2875800e565616822a03f"}]},"branch":"refs/heads/master","description":"Rebase"},"8ad82835fa5a8e6448b2c92cd3050cb4b53200f1":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-01-09 10:53:00.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/17"}}},"commit":{"parents":[{"commit":"f0c0ff4efdf63fb4b73bd77d678467e9aeac1619","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0c0ff4efdf63fb4b73bd77d678467e9aeac1619"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-09 10:53:00.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ad82835fa5a8e6448b2c92cd3050cb4b53200f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ad82835fa5a8e6448b2c92cd3050cb4b53200f1"}]},"branch":"refs/heads/master","description":"Rebase"},"04a4562fe7dff0ff3bbc29eb33123c095331371b":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-01-16 14:17:52.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/18"}}},"commit":{"parents":[{"commit":"978049fa158f1d844edf69e4487d20f4c3e4bef5","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/978049fa158f1d844edf69e4487d20f4c3e4bef5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-16 14:17:52.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04a4562fe7dff0ff3bbc29eb33123c095331371b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/04a4562fe7dff0ff3bbc29eb33123c095331371b"}]},"branch":"refs/heads/master","description":"Rebase"},"b0852562a56a11a3df9747aa57c68087be200916":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-01-22 14:25:27.000000000","uploader":{"_account_id":33634,"name":"Jan Hartkopf","email":"j@hartkopf.io","username":"jhartkopf"},"ref":"refs/changes/14/892814/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/19"}}},"commit":{"parents":[{"commit":"60a6ceed09515842366e67de86b5aef783f1dcd2","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60a6ceed09515842366e67de86b5aef783f1dcd2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Jan Hartkopf","email":"jhartkopf@inovex.de","date":"2024-01-22 14:25:27.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch addes the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/900870\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0852562a56a11a3df9747aa57c68087be200916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b0852562a56a11a3df9747aa57c68087be200916"}]},"branch":"refs/heads/master","description":"Rebase"},"0df02d5cb2a12343ef1c6b33c2f8b690b52a3f45":{"kind":"REWORK","_number":20,"created":"2024-02-19 16:40:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/14/892814/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/20"}}},"commit":{"parents":[{"commit":"60a6ceed09515842366e67de86b5aef783f1dcd2","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60a6ceed09515842366e67de86b5aef783f1dcd2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-19 16:40:17.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch adds the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nIt also adds a test job \"designate-bind9-catalog-zones\" that uses catalog zones\nwith a BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0df02d5cb2a12343ef1c6b33c2f8b690b52a3f45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0df02d5cb2a12343ef1c6b33c2f8b690b52a3f45"}]},"branch":"refs/heads/master"},"2b93f4435f6273e073d14795ef66aa977aa38e6d":{"kind":"REWORK","_number":21,"created":"2024-02-20 15:10:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/14/892814/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/21"}}},"commit":{"parents":[{"commit":"60a6ceed09515842366e67de86b5aef783f1dcd2","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60a6ceed09515842366e67de86b5aef783f1dcd2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-20 15:10:13.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch adds the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nIt also adds a test job \"designate-bind9-catalog-zones\" that uses catalog zones\nwith a BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b93f4435f6273e073d14795ef66aa977aa38e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b93f4435f6273e073d14795ef66aa977aa38e6d"}]},"branch":"refs/heads/master"},"4258633ae5713b2dbec3bfe1a047bc76fd7a10f0":{"kind":"REWORK","_number":22,"created":"2024-02-20 18:24:43.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/14/892814/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/14/892814/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/14/892814/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/14/892814/22"}}},"commit":{"parents":[{"commit":"60a6ceed09515842366e67de86b5aef783f1dcd2","subject":"Add support for catalog zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60a6ceed09515842366e67de86b5aef783f1dcd2"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-08-25 20:14:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-20 18:24:36.000000000","tz":0},"subject":"Add catalog zone support to devstack plugin","message":"Add catalog zone support to devstack plugin\n\nThis patch adds the ability to configure devstack to use catalog zones\nwhen using the BIND9 backend.\n\nIt also adds a test job \"designate-bind9-catalog-zones\" that uses catalog zones\nwith a BIND9 backend.\n\nChange-Id: Ib618d7850b0a86a8eb10eaa52b0e330cf908883a\nSigned-off-by: Jan Hartkopf \u003cjhartkopf@inovex.de\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4258633ae5713b2dbec3bfe1a047bc76fd7a10f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4258633ae5713b2dbec3bfe1a047bc76fd7a10f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
