)]}'
{"id":"openstack%2Fdesignate~896061","triplet_id":"openstack%2Fdesignate~master~Ie0419cdb52dbfcb813faf9070110a2635a46ed51","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-09-26 19:55:38.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ie0419cdb52dbfcb813faf9070110a2635a46ed51","subject":"Prevent XFR from being used on PRIMARY zones","status":"MERGED","created":"2023-09-21 12:05:19.000000000","updated":"2023-09-26 21:01:28.000000000","submitted":"2023-09-26 21:00:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896061","meta_rev_id":"6164177901aac992d2dd31f47897fb103e0431cb","_number":896061,"virtual_id_number":896061,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 21:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-26 15:19:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 19:55:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 13:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 15:19:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"34759048c950fcc0015f97704bf07d667c5be65c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:05:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f39f736f7281df11f7c860340413a419b59cf460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:37:20.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"dc80b9ad61ee047568c05cc712c48c08e9821738","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:37:31.000000000","message":"Topic safety removed","accounts_in_message":[],"_revision_number":2},{"id":"2685c43d6a327d7cc0b6178c8ca756511e632a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:42:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05b943f89c294669b849beba7363fb90\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16476b3251b54d41ba4d4375f3fe9574 : SUCCESS in 41m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/271f0802c685493b8a88e1f70ed698c8 : SUCCESS in 43m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fbb8b8b85b941dba84de5a14ccd9988 : SUCCESS in 37m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f08b201d99df49f8825f9b7c9e0388cc : SUCCESS in 36m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f6181343fa6c48b8907d7c1f86c78e94 : SUCCESS in 41m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04cff261a461436a80619e21c94f93c0 : SUCCESS in 56m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b5343a533a19488c8a59c906db854af1 : SUCCESS in 1h 01m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e586a37e87a40eeab37dba572204fb9 : SUCCESS in 38m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d86ff12b4d014aaaafd25a0b4919f0f2 : SUCCESS in 39m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3516e074655447a2ace43d6dd85e9765 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/426c0bd7178c4dfd907d5ab9be14d94b : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3c0e58bab3b4b7daf0bfc4d0e2ed1d8 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85be9baf87c14fccaed035e10a3045f1 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e8fc9d62eac467494769d99fb010a82 : FAILURE in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/258f964dc42d414f92e3ef22e03f5986 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5fa5d5f6c04174949afc15e2eaca22 : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14f263657c0e42e3bdb402c2cb24f80e : SUCCESS in 41m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/94464169fe2a426ca0a85ca7fd6594e8 : SUCCESS in 6m 56s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e9cd9b68239b42b780284b623d1755d1 : SUCCESS in 46m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/30d92f18072f4eac8138e50d6d47bb4b : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a2ab2330b7fdeada4393e4bb355695f9c1b4ab87","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 14:17:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bd5e1fbbabd637f4b9d94b3868467658b39fa6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:53:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f285646c8c245d28358f25af889aa27\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5e9a8a4e34fe44eb81e24efeaef313a5 : SUCCESS in 46m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4cbc0f4a33c8439aadee37d9fda9e5c2 : SUCCESS in 1h 05m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b105548128d4b7ebc27decc5188182f : SUCCESS in 59m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/480c35df059d4bb39d2699e89dfdf5b9 : SUCCESS in 50m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1d01f3e2ff4b45788e69b78eb72274e9 : SUCCESS in 37m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/753730da90454b50bf1475c622ef411b : SUCCESS in 1h 28m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32ed99483b994c76aae1f329c41b6c90 : SUCCESS in 1h 04m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a024636c57934910bf0f39e80b06230c : SUCCESS in 31m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0cdc6fdb4b074c27a474dcc99331a657 : SUCCESS in 48m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4da1dcb51237453795d841d8725be8cb : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed360585b574220984bf14a6cfc9eeb : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e1e1c6c7e9642e08ff8d61042effec1 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8860aeba80347f5bf3b9741e5bac9fc : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61ee4eb5387141c4b2a238b629eb94a6 : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8137cc9d565245be9826101e73f719c2 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea8ff6f05ed047cf9e57caea7c5ca4b5 : SUCCESS in 8m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ab81acac5ca64e53a9e1071259bb2e51 : SUCCESS in 35m 12s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/647b2d2006534ee79502b3b9ed2ef8c7 : SUCCESS in 7m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5d0941e5ded249b488332aed6c23254f : SUCCESS in 44m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/18ec9f407a3d4663a1bac5991b2c1029 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"51635111f019c9b595ce6ae343091ae323fdb3e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-26 15:19:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3b1fc56cf77826978431c9ecae53d26612ccf19","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-26 19:55:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b9a2032f22d2835cec6a1255eaff3b6d0a7f7d51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 19:56:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b110c32f494f4bf2cbf7f2bcc9c9ff0dfb641acb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 21:00:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe761d5f278a401eb63c9f2789aa80b9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a33c7a157d874517ba8394b54e1c3665 : SUCCESS in 34m 32s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/45c0cab3c5c04ac4882c9be7d8559906 : SUCCESS in 33m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/932b3c6a9dba4b47bf849303d1ba1606 : SUCCESS in 34m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/627e2d3a65224b439ef532f62dd1901a : SUCCESS in 55m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e050f8e61af348b1837ef8e1323251c1 : SUCCESS in 44m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/51d481a0e6164b33b7b73c5a663bbedc : SUCCESS in 47m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4df547e374154b1bb31379aa0b2445df : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdd0ab650649430fb6d4417e85e11658 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f8da44d1313493897a7dfd1aa63be0c : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d510b83f9e4a72b9cffe8daf803b2c : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc66ef485bc3462eb06d1d1f4625df55 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b7710c652b9d40a28d30bdbe03a80474 : SUCCESS in 43m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cf6fa5f074314287b5b8a43a834a8b76 : SUCCESS in 7m 10s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/329357d6ec31464fb015bde5303ea2ae : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":2},{"id":"fd73944d75026c7a05a3ab238b69e65c1229897c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 21:00:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6164177901aac992d2dd31f47897fb103e0431cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 21:01:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f88bfc8fad4faf8ede81faa736e2cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ff6b9e11cf4623a944e1032573fbfb : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d","revisions":{"6cf3323513c11fe867d39ac648650b0f2fd3e24a":{"kind":"REWORK","_number":1,"created":"2023-09-21 12:05:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/61/896061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/896061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/896061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/896061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/896061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/896061/1"}}},"commit":{"parents":[{"commit":"aaed97875c870d4807bcb6694345ea2e80f437f8","subject":"Fix bug with designate-manage pool update losing existing record data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aaed97875c870d4807bcb6694345ea2e80f437f8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:04:41.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:04:41.000000000","tz":-420},"subject":"Prevent XFR from being used on PRIMARY zones","message":"Prevent XFR from being used on PRIMARY zones\n\n- Added constant for PRIMARY / SECONDARY.\n\nChange-Id: Ie0419cdb52dbfcb813faf9070110a2635a46ed51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cf3323513c11fe867d39ac648650b0f2fd3e24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cf3323513c11fe867d39ac648650b0f2fd3e24a"}]},"branch":"refs/heads/master"},"79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-21 12:37:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/61/896061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/896061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/896061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/896061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/896061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/896061/2"}}},"commit":{"parents":[{"commit":"aaed97875c870d4807bcb6694345ea2e80f437f8","subject":"Fix bug with designate-manage pool update losing existing record data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aaed97875c870d4807bcb6694345ea2e80f437f8"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:04:41.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:37:20.000000000","tz":0},"subject":"Prevent XFR from being used on PRIMARY zones","message":"Prevent XFR from being used on PRIMARY zones\n\n- Added constant for PRIMARY / SECONDARY.\n\nRelated-Bug: #2036750\nChange-Id: Ie0419cdb52dbfcb813faf9070110a2635a46ed51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
