)]}'
{"id":"openstack%2Fdesignate~896062","triplet_id":"openstack%2Fdesignate~master~Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-09-26 22:15:31.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272","subject":"Fix AXFR loop when updating secondary zone","status":"MERGED","created":"2023-09-21 12:30:52.000000000","updated":"2023-09-26 23:29:36.000000000","submitted":"2023-09-26 23:28:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896062","meta_rev_id":"bd91a4ef56f0631dcd2cd1d54bad797818e4cbea","_number":896062,"virtual_id_number":896062,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 23:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 21:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-09-26 22:15:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 13:39:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-26 21:19:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"014b2efc7ba52f9461d116d0aeb2895e239c65ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:30:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc8ed44600f7b913485330bc6b07da7c7f304764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:32:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"040208918056a9c31bb085542d97088ce9402741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:37:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"a4ffd49e5151803e46d83a929a74aebeec17904a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:37:30.000000000","message":"Topic loop removed","accounts_in_message":[],"_revision_number":3},{"id":"e86b81ec6db9e0797d1bf5b94d37c4774d5de13d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:39:02.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c4a575433db96a5c97abd829f30b289a250ea203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:39:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c1592cf72134138bd02112a51b07191\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0169413be6a48f09b2ee83e7d8c8c5e : SUCCESS in 41m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/01c8ec362dc941bd8142cbfee27405b9 : SUCCESS in 47m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1e84763ccda475ebfbb3a2fa27e05cd : SUCCESS in 40m 26s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0226a190a2ca496f9d62113d24a2f25f : SUCCESS in 49m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a0cc1095ef6247c68d53d412793ea0ea : SUCCESS in 52m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0974abc42a2a49ad9efc57f704a23899 : SUCCESS in 56m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1351a01e78c04b9d90ea9e1d7a7fedbc : SUCCESS in 58m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f8e3c23688af49ac953cd163df20887d : SUCCESS in 36m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/638db70000314c6288485a56b691346a : SUCCESS in 40m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/819f1fe791aa4c13ac67aca14c3c0118 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/582944ce992641028d4d10039f92f753 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f4247ac0afc4de5801b76249a25b63b : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19f1422dee3045bd94c9f65a9f73f9bd : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93d8263b278c4402837e7578ee83fccd : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/622397989ca54620a9e2160b648ea4d3 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1f1f67a294444cabb1f7e7f606b790 : SUCCESS in 8m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cfa090494d4c45ecaabc94508ede85ae : SUCCESS in 32m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f7becfb2ac9745fc96753c5b61ca820c : SUCCESS in 6m 17s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4fa5a5c68a2b46268828cdbf14ee8d9f : SUCCESS in 39m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dfc16fd8f0c6406798168487ebff926d : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"52dcb06803c7f31904d3194ba2a02bef130b51c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-26 19:32:30.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7c2a506bbb202f6785ca7fa36efe5f4e83accb41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-26 19:38:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ee65cf562e69e48dfae96c10660f4778deacc6a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 20:46:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c48880878e0d4e5bba5fc63f95f27615\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/909b9d5471ad463b862d3b823aa79382 : SUCCESS in 54m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/280f052932ba43d387d3e38c8063c0a1 : SUCCESS in 1h 03m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/97a9b34d3b8c423cb3ade731e6e7e5f1 : SUCCESS in 58m 41s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1f1a1f0348fd410886904833ff1c3b6d : SUCCESS in 50m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2264aa75a2aa40968e11b78cbce53fa0 : SUCCESS in 47m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eafe76914b554b41b0705ad9bc97545d : SUCCESS in 56m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d5fe3cd8be524857ae3cd3f7f7f0989a : SUCCESS in 1h 04m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f967bcda083746d38d40c18aa184a4d2 : SUCCESS in 43m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec3aab419d7a474c84cb340d08ddf23a : SUCCESS in 45m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ecdc02fd308408fbc05f472f3336c64 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb2e3f83c92464082d33cdce8924a40 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1317ad1539c7497aa77216a1b2562b3c : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1b5038c3eeb4f0aa4d63aa3bdba544d : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9e1f3050a934a6a8fbc74d8e981a38e : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5277dd750eb349ca94e35bed63a952ce : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a8582bc1084e239b7fbf97a7371b64 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0836be4be56f4cf89ee65af57a50da54 : SUCCESS in 30m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f65fa70d17e64e22b7a61a4aa8d6ff93 : SUCCESS in 5m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0185b2c62eee45f1b66d19dd27cd8308 : SUCCESS in 36m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ad01362cd19240b4b49e8f6737e4ff5f : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1a69368c71affe9cf6319cd94fa46f20ccff3333","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-26 21:19:46.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"67df9280f635c9dd8578f7b6a9f9265ca144f248","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-26 22:15:31.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"a32d7221cefb529350bdbd8e46809bf724dd6ad7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 22:16:05.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f52ede2794e588a68a11dd507caeb0603c62ef6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 23:28:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279e6fb2f27b467a851e310ea7e4084f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bf8036aa3ea2459dbca4a9d5fc2da4f5 : SUCCESS in 47m 21s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f851d767672740d1bd280c1aa5f437f0 : SUCCESS in 30m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eae6ef43f87f438893c960cc146a0213 : SUCCESS in 46m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1ad3fc89bb104ec5bc43688db5438c8c : SUCCESS in 1h 07m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4803b0ad9ff242678e4c7c0d4b3f0791 : SUCCESS in 48m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ce5ed63f4d741f192999e4db4b71fc8 : SUCCESS in 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21965a981d864db0b83524951308c06e : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b92bfb304ff432db638eec0532c58a5 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19dde4db420b45dc94d43eb88533dc2f : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29c2e98c2443432c865307c7d8112a06 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa2e3761fed4d55a48f9d9149edc424 : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/feb28303f00d48c7bed709c0fca96c02 : SUCCESS in 38m 23s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/768c9e7730e14df1b99219cfadbfb764 : SUCCESS in 5m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0a7bab0010d940228d05692217b5b476 : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":6},{"id":"17b92529c8dd34d3bb875c7d4ada04fb814df8a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 23:28:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"bd91a4ef56f0631dcd2cd1d54bad797818e4cbea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 23:29:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/259147e5d6ff4ffe86f29e7957740af5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78a859bfc7b42a3b6883f3324b37036 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ecaff9e6d348ce3b47a4b4fb71896951274191cb","revisions":{"f046199f8ceca4fc1df230860fe6af43e7554b80":{"kind":"REWORK","_number":1,"created":"2023-09-21 12:30:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/896062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/896062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/896062/1"}}},"commit":{"parents":[{"commit":"6cf3323513c11fe867d39ac648650b0f2fd3e24a","subject":"Prevent XFR from being used on PRIMARY zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cf3323513c11fe867d39ac648650b0f2fd3e24a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:28:45.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:29:52.000000000","tz":-420},"subject":"Fix AXFR loop when updating secondary zone","message":"Fix AXFR loop when updating secondary zone\n\nWe previously fixed this loop when creating a secondary\nzone, but the same thing happens when you update a zone.\n\nThis patch moves the fix to the AXFR code to prevent all\nsecondary zone AXFR loops.\n\nPartial-Bug: #1856442\nChange-Id: Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f046199f8ceca4fc1df230860fe6af43e7554b80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f046199f8ceca4fc1df230860fe6af43e7554b80"}]},"branch":"refs/heads/master"},"6781af1f57da47fc9337dc305f8108b631140e7c":{"kind":"REWORK","_number":2,"created":"2023-09-21 12:32:13.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/896062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/896062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/896062/2"}}},"commit":{"parents":[{"commit":"6cf3323513c11fe867d39ac648650b0f2fd3e24a","subject":"Prevent XFR from being used on PRIMARY zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6cf3323513c11fe867d39ac648650b0f2fd3e24a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:28:45.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:32:06.000000000","tz":-420},"subject":"Fix AXFR loop when updating secondary zone","message":"Fix AXFR loop when updating secondary zone\n\nWe previously fixed this loop when creating a secondary\nzone, but the same thing happens when you update a zone.\n\nThis patch moves the fix to the AXFR code to prevent all\nsecondary zone AXFR loops.\n\nPartial-Bug: #1856442\nChange-Id: Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6781af1f57da47fc9337dc305f8108b631140e7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6781af1f57da47fc9337dc305f8108b631140e7c"}]},"branch":"refs/heads/master"},"b07cf8d4d62009487f6cf527ebbbb741c4459554":{"kind":"NO_CHANGE","_number":3,"created":"2023-09-21 12:37:26.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/896062/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/896062/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/896062/3"}}},"commit":{"parents":[{"commit":"79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d","subject":"Prevent XFR from being used on PRIMARY zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:28:45.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:37:26.000000000","tz":0},"subject":"Fix AXFR loop when updating secondary zone","message":"Fix AXFR loop when updating secondary zone\n\nWe previously fixed this loop when creating a secondary\nzone, but the same thing happens when you update a zone.\n\nThis patch moves the fix to the AXFR code to prevent all\nsecondary zone AXFR loops.\n\nPartial-Bug: #1856442\nChange-Id: Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b07cf8d4d62009487f6cf527ebbbb741c4459554"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b07cf8d4d62009487f6cf527ebbbb741c4459554"}]},"branch":"refs/heads/master","description":"Rebase"},"8aada468b5d297c6d4c44af07bd5213aa5a3967c":{"kind":"REWORK","_number":4,"created":"2023-09-21 12:39:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/896062/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/896062/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/896062/4"}}},"commit":{"parents":[{"commit":"79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d","subject":"Prevent XFR from being used on PRIMARY zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:28:45.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:39:02.000000000","tz":0},"subject":"Fix AXFR loop when updating secondary zone","message":"Fix AXFR loop when updating secondary zone\n\nWe previously fixed this loop when creating a secondary\nzone, but the same thing happens when you update a zone.\n\nThis patch moves the fix to the AXFR code to prevent all\nsecondary zone AXFR loops.\n\nPartial-Bug: #1856442\nChange-Id: Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8aada468b5d297c6d4c44af07bd5213aa5a3967c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8aada468b5d297c6d4c44af07bd5213aa5a3967c"}]},"branch":"refs/heads/master"},"549cdb937ca3910ecf3b8af60b31ad5c9022e8e2":{"kind":"REWORK","_number":5,"created":"2023-09-26 19:32:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/896062/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/896062/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/896062/5"}}},"commit":{"parents":[{"commit":"79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d","subject":"Prevent XFR from being used on PRIMARY zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:28:45.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-26 19:29:39.000000000","tz":-420},"subject":"Fix AXFR loop when updating secondary zone","message":"Fix AXFR loop when updating secondary zone\n\nWhen a secondary zone is created or updated we trigger an AXFR,\nthis AXFR then goes ahead and updates the zone. This then causes\nthe AXFR code to get triggered again as it thinks the masters\nhas changed.\n\nWe fix this by ensuring that when we update the zone from the AXFR\ncode that we always reset the masters field.\n\nPartial-Bug: #1856442\nChange-Id: Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/549cdb937ca3910ecf3b8af60b31ad5c9022e8e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/549cdb937ca3910ecf3b8af60b31ad5c9022e8e2"}]},"branch":"refs/heads/master"},"ecaff9e6d348ce3b47a4b4fb71896951274191cb":{"kind":"REWORK","_number":6,"created":"2023-09-26 19:38:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/62/896062/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/62/896062/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/62/896062/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/62/896062/6"}}},"commit":{"parents":[{"commit":"79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d","subject":"Prevent XFR from being used on PRIMARY zones","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79aac2b2062f0379eaa7b105f5722bd4cbf0ad8d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:28:45.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-26 19:38:04.000000000","tz":-420},"subject":"Fix AXFR loop when updating secondary zone","message":"Fix AXFR loop when updating secondary zone\n\nWhen a secondary zone is created or updated we trigger an AXFR,\nthis AXFR then goes ahead and updates the zone. This then causes\nthe AXFR code to get triggered again as it thinks the masters\nhas changed.\n\nWe fix this by ensuring that when we update the zone from the AXFR\ncode that we always reset the masters field.\n\nPartial-Bug: #1856442\nChange-Id: Ifc2efb5a6d1091ff16373f3cbe3846f5e4371272\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ecaff9e6d348ce3b47a4b4fb71896951274191cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ecaff9e6d348ce3b47a4b4fb71896951274191cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
