)]}'
{"id":"openstack%2Fdesignate~896103","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I77a5d05cd8f55e975a16cb386221d53b63131886","project":"openstack/designate","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-09-27 22:45:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I77a5d05cd8f55e975a16cb386221d53b63131886","subject":"Fix bug with designate-manage pool update losing existing record data","status":"MERGED","created":"2023-09-21 12:58:07.000000000","updated":"2023-09-28 00:01:16.000000000","submitted":"2023-09-28 00:00:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896103","cherry_pick_of_change":896065,"cherry_pick_of_patch_set":1,"meta_rev_id":"6fd515c62edcb5e9a2bb3c2f569641008322b650","_number":896103,"virtual_id_number":896103,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 00:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-21 16:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-27 22:45:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 14:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 16:07:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3f3e5ab3da106e3158f5e7349a2b935f493e0bf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:58:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"93fed0f8c2505b6c5f6f5e14a80d311d81f50b6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 14:04:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61d23e3ebcad4c47b70bdb6ff8c5a5a5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d26047cc1f7f4b09850dd1c4c5e4f61e : SUCCESS in 39m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f6325a00734b4122a8671c5491875b26 : SUCCESS in 43m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d5fe883b6abe476f95756d7ea321d7b1 : SUCCESS in 1h 05m 05s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab6120cdace1408f9b380c93564dd18b : SUCCESS in 44m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4f9972326f724c5ab484a96a8b73c6ad : SUCCESS in 42m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d2566c92ac643e399c0e609c43d8cce : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0554b2fd92d0422385c95190c62a605b : FAILURE in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90fce548c8e24fc7896783a6dae7b277 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/976fe3901150402db83b3c350f6f6609 : FAILURE in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa07da703c84832ac8d2707e9ab998a : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/6fd9541d6c7e4a9e84e1102c7cae653f : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":1},{"id":"72a032080e25284760d41d5994ca24e58aa42301","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 14:13:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bafdc1da82dc1ff4b66d9acb0eed2eac7b18b65a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:16:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca64947b1da049bca7a6c45d630415a5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/31a95a7951fb4282bf9b570fc262a72d : SUCCESS in 42m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/42b0fe15104945189e1313858a798da2 : SUCCESS in 38m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/71cbadd65c07436bb62536acf3b8f7fb : FAILURE in 24m 30s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d942f11e5e93422ab243f2d96945d8e9 : SUCCESS in 48m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1d765ae9863b4af195e0415553caa4ed : SUCCESS in 40m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/381f3f97565b4405a721dbd5e2b47076 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86fbc7eb5b64fc5b8018ea72a2762a2 : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044f64275ed34421bfbc127cd14461af : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39687bbbc75d4e13997f0b87416c6777 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5028f2aa0a7a41cbaac544b71e6feb9f : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/34bf105fc7e2429bbabc291cfe747114 : SUCCESS in 55m 33s","accounts_in_message":[],"_revision_number":2},{"id":"00dc5bb7b7288261de77c12499ea4321b0f94952","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-21 16:07:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cb26886147d6d6b3f7b865e0b4c2a6873ef688b4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-27 22:45:53.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"75c73455e345de4251fe0d7c60f0f2d32ccec28b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 22:46:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f71673c67f28588b6b9cacd94aded5dfcd27fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 00:00:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55bf94ec4ea14682bab750c4bd32c434\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5d1e427df897492bb6e59aabd51914bc : SUCCESS in 35m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/49e6919a1f8d4582a88a5b65115b5a2d : SUCCESS in 44m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/23356f8b71124470ab0a242628df63cd : SUCCESS in 42m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b0f60e4721fc40a9840e2a781de62cb2 : SUCCESS in 30m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e121eef28349ec8103b7c2bc1e31d0 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/006e40299c7d401aaafeaff8881de11c : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/310623ac452a4ac9ab3510845ef48e41 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dceb1aa6f7774e51b95712ab793e66cd : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":2},{"id":"d503b33a934890477275faf90b5ef306b5176625","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 00:00:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6fd515c62edcb5e9a2bb3c2f569641008322b650","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 00:01:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/866f823490c34d65843d0709fba577eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea7aaaa08ce4142b572980ffafe0261 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"de6715d961becd6962c7d8f8def6c2eda9c7351c","revisions":{"4d564931d42f6bb717f69b5f46238dc021b70652":{"kind":"REWORK","_number":1,"created":"2023-09-21 12:58:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/896103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/896103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/896103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/896103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/896103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/896103/1"}}},"commit":{"parents":[{"commit":"4673e091fe9fcc1d6154f6430c74f8e4f56203e5","subject":"[stable-only] Cap virtualenv/setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4673e091fe9fcc1d6154f6430c74f8e4f56203e5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-20 21:49:53.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:58:07.000000000","tz":0},"subject":"Fix bug with designate-manage pool update losing existing record data","message":"Fix bug with designate-manage pool update losing existing record data\n\nWhen designate-manage pool update is used we trigger an AFXR and\nuse that information to rebuild the zone. This is however only valid\nfor secondary zones, as the AXFR does not contain all the internal\ninformation required for a primary zone and that causes designate to\nlose important internal record data.\n\nCloses-Bug: #2036750\nChange-Id: I77a5d05cd8f55e975a16cb386221d53b63131886\n(cherry picked from commit aaed97875c870d4807bcb6694345ea2e80f437f8)\n(cherry picked from commit f2624f73cf3b9b95b7342ca8c89ef9b8e7e438e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d564931d42f6bb717f69b5f46238dc021b70652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d564931d42f6bb717f69b5f46238dc021b70652"}]},"branch":"refs/heads/stable/victoria"},"de6715d961becd6962c7d8f8def6c2eda9c7351c":{"kind":"REWORK","_number":2,"created":"2023-09-21 14:13:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/896103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/896103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/896103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/896103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/896103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/896103/2"}}},"commit":{"parents":[{"commit":"4673e091fe9fcc1d6154f6430c74f8e4f56203e5","subject":"[stable-only] Cap virtualenv/setuptools","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4673e091fe9fcc1d6154f6430c74f8e4f56203e5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-20 21:49:53.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 14:13:20.000000000","tz":-420},"subject":"Fix bug with designate-manage pool update losing existing record data","message":"Fix bug with designate-manage pool update losing existing record data\n\nWhen designate-manage pool update is used we trigger an AFXR and\nuse that information to rebuild the zone. This is however only valid\nfor secondary zones, as the AXFR does not contain all the internal\ninformation required for a primary zone and that causes designate to\nlose important internal record data.\n\nCloses-Bug: #2036750\nChange-Id: I77a5d05cd8f55e975a16cb386221d53b63131886\n(cherry picked from commit aaed97875c870d4807bcb6694345ea2e80f437f8)\n(cherry picked from commit f2624f73cf3b9b95b7342ca8c89ef9b8e7e438e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de6715d961becd6962c7d8f8def6c2eda9c7351c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de6715d961becd6962c7d8f8def6c2eda9c7351c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
