)]}'
{"id":"openstack%2Fdesignate~896104","triplet_id":"openstack%2Fdesignate~stable%2Fussuri~I77a5d05cd8f55e975a16cb386221d53b63131886","project":"openstack/designate","branch":"stable/ussuri","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-09-28 14:26:50.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I77a5d05cd8f55e975a16cb386221d53b63131886","subject":"Fix bug with designate-manage pool update losing existing record data","status":"MERGED","created":"2023-09-21 12:58:15.000000000","updated":"2023-09-28 15:40:27.000000000","submitted":"2023-09-28 15:39:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896104","cherry_pick_of_change":896065,"cherry_pick_of_patch_set":1,"meta_rev_id":"cdd8d8680e53210dd3c0ff8da5d548f43bdb7040","_number":896104,"virtual_id_number":896104,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-28 15:39:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-21 16:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-28 14:26:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 15:33:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 16:08:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"fbf14bb60c9d81143308498da9a0d7bf918a3d16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 12:58:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"1f2cdda9e663c4793ec9e635296d97a75f0f6ebc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 14:13:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23766f00d8f4977826792607fd435977de07515d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:33:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c724c5c65a9345fbac8544f8579a64ff\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a0dcadcf0bf41bea0aadc29d40c3587 : SUCCESS in 33m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/683d7edc25214457b94cb7bd389c805a : SUCCESS in 39m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/16ad3922d87e4c5d856bd649f75b4a89 : SUCCESS in 1h 13m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/37fee54899da469b9809243095d39f46 : SUCCESS in 56m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c2dfdb6a9734e8396e1bc4f834af97f : SUCCESS in 39m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/700b454c74a341a88bfd9234b1a63c0e : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc46b54219d4478821119656406069e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea8bb95c1f3b47df8f024476ac7b5506 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c9f06a7d5344d1fa81e6df55d19934a : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd16eb698c44cb283584139a7889da3 : SUCCESS in 6m 37s\n- neutron-tempest-plugin-designate-scenario-ussuri https://zuul.opendev.org/t/openstack/build/47891f15e1454a5cb61eca7922797ffe : SUCCESS in 32m 25s","accounts_in_message":[],"_revision_number":2},{"id":"0c5d5da35a38d3ec189dd90f7c8368a340cf342f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-21 16:08:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"73c0f0cb410f95d5a9b43955d079b7d4418731be","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-28 14:26:50.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac840980753d9fba298f561ddc89fa39cabd7fa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 14:27:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c866d2d33f56d1c4519979d2674d08c68e1e460","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:39:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a800804702064b5e9b4294b81e43f409\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40f42193d482459cb6a4b78b1f4862e4 : SUCCESS in 38m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5e2a90f49f694707b9873da9e09703a4 : SUCCESS in 34m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4de20322f4ba4946aadf6deb3471fae5 : SUCCESS in 1h 11m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d4fdc3921ce14966ad6bcad7c8509403 : SUCCESS in 45m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d7577cd9be4c4395a8e7f54d4794de53 : SUCCESS in 32m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac853d401474b059b48bafd8e9c1158 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b9c0e7aa47a4552a0ace50d7a148062 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0f42961c2b5463fb4bf83f8911d6895 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aeb0c6ffde94a43ba3d0244e8f13022 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":2},{"id":"12514d5c6a50dae0cf577d8c8279285f5c6984f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:39:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cdd8d8680e53210dd3c0ff8da5d548f43bdb7040","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:40:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d09a3672c7c4e6f94d401464cf75f0f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab0b6147412403e8460280884080264 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a838ecbe0febceec1408f24735f3bc06d283054","revisions":{"656f56cd302a9fc71187e3d11441a08574c9033d":{"kind":"REWORK","_number":1,"created":"2023-09-21 12:58:15.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/896104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/896104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/896104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/896104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/896104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/896104/1"}}},"commit":{"parents":[{"commit":"dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-20 21:49:53.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 12:58:15.000000000","tz":0},"subject":"Fix bug with designate-manage pool update losing existing record data","message":"Fix bug with designate-manage pool update losing existing record data\n\nWhen designate-manage pool update is used we trigger an AFXR and\nuse that information to rebuild the zone. This is however only valid\nfor secondary zones, as the AXFR does not contain all the internal\ninformation required for a primary zone and that causes designate to\nlose important internal record data.\n\nCloses-Bug: #2036750\nChange-Id: I77a5d05cd8f55e975a16cb386221d53b63131886\n(cherry picked from commit aaed97875c870d4807bcb6694345ea2e80f437f8)\n(cherry picked from commit f2624f73cf3b9b95b7342ca8c89ef9b8e7e438e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656f56cd302a9fc71187e3d11441a08574c9033d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/656f56cd302a9fc71187e3d11441a08574c9033d"}]},"branch":"refs/heads/stable/ussuri"},"9a838ecbe0febceec1408f24735f3bc06d283054":{"kind":"REWORK","_number":2,"created":"2023-09-21 14:13:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/896104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/896104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/896104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/896104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/896104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/896104/2"}}},"commit":{"parents":[{"commit":"dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a","subject":"Removed unsupported test B309 from bandit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dcb5d6ca0f00a5d0f487ac6e0dbbf0b5ddf26b1a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-20 21:49:53.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 14:13:53.000000000","tz":-420},"subject":"Fix bug with designate-manage pool update losing existing record data","message":"Fix bug with designate-manage pool update losing existing record data\n\nWhen designate-manage pool update is used we trigger an AFXR and\nuse that information to rebuild the zone. This is however only valid\nfor secondary zones, as the AXFR does not contain all the internal\ninformation required for a primary zone and that causes designate to\nlose important internal record data.\n\nCloses-Bug: #2036750\nChange-Id: I77a5d05cd8f55e975a16cb386221d53b63131886\n(cherry picked from commit aaed97875c870d4807bcb6694345ea2e80f437f8)\n(cherry picked from commit f2624f73cf3b9b95b7342ca8c89ef9b8e7e438e4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a838ecbe0febceec1408f24735f3bc06d283054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a838ecbe0febceec1408f24735f3bc06d283054"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
