)]}'
{"id":"openstack%2Fdesignate~896118","triplet_id":"openstack%2Fdesignate~stable%2Fyoga~I677ac79047286fb92343d3d895c38cf1a1076556","project":"openstack/designate","branch":"stable/yoga","topic":"cinder-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-09-25 22:54:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I677ac79047286fb92343d3d895c38cf1a1076556","subject":"[CI] Reduce workers and disable cinder to save on memory","status":"MERGED","created":"2023-09-21 22:42:51.000000000","updated":"2023-09-25 22:56:28.000000000","submitted":"2023-09-25 22:54:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896118-cinder-stable/zed-stable/yoga","cherry_pick_of_change":874840,"cherry_pick_of_patch_set":2,"meta_rev_id":"f792a0e13dd029197735497dd74b529e761c8e28","_number":896118,"virtual_id_number":896118,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-25 22:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-25 21:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-09-25 21:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 23:45:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 21:46:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a6708e19bbc7c99926762ec219828ea40873bdc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-09-21 22:42:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"5f5d15c436361b4a2652b3c1b6bd7d314b037f1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 23:45:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/030d9fe591d94beb9b0c15dc60014cd7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f506fc9b5fb74e6799dad16d55d7a7e0 : SUCCESS in 51m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a5ed912fa24f4f50941ff82092ccc372 : SUCCESS in 57m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/af78b99d95494d8787fcb317d29545f2 : SUCCESS in 52m 27s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/833c3794630b426b9f73de6b3c9be69d : SUCCESS in 50m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45180b55b8d046be89bae2262e9ab7dc : SUCCESS in 27m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/609ea7d9f34140c58052bf8bddb5012e : SUCCESS in 39m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c096c1ac4ff4d5eb710d84d9f6e0ecd : SUCCESS in 36m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e8142772b970402fba7bf596d526af82 : SUCCESS in 29m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b7f28bd35f1469f8a951b7e3e28b7a6 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30ad4173d8514f189d8573fbf65a0fe0 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6ce8b7c348b418dbba9ad99a2265417 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5566c794c29b4151b6c07aa345eaef24 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c05124afad7463796e8444ab2396a60 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/df8b6cfc8e6a4764acb31fb423981b88 : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f7520ab2d0118276062222e178e78ac4064278a4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-25 21:46:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e22f54d06e2b53896b62c919cc2e03d63a1db5a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 21:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca15d8ac887fc8b07e08359e38c99f9ebbda6bc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 22:54:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7ebc5882224407bbc5b5add0a99a003\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/963566c743104841b98b7ce157877dd8 : SUCCESS in 46m 35s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/0e4ddf55763b47aba4f2dfc108f552b6 : SUCCESS in 56m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b2ddf8b7499f4585aa8861dc15427bfc : SUCCESS in 41m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/403fa90496474f39a914493b5d5afc0b : SUCCESS in 1h 03m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cf53c133467f4d92a06bfbdd53b9a084 : SUCCESS in 51m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cbacdc08946c4521873a0f58ac834c8f : SUCCESS in 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d455c1294242d5b5b3e5b0a5ef38cf : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/891acb5693134c46b63e974a72ef9639 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b88e146d55f4b4c9cbc216136b2de2d : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e61995ea9d44b196b049ac2d92737a : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2efab44cc93e4e7aab2ba2cdde566b55 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":1},{"id":"dff59c2a52c0aab53b66503fbc1bf9f81d4d0563","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 22:54:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f792a0e13dd029197735497dd74b529e761c8e28","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 22:56:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab3c3e0d33947c1815abf9c4a594b6d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772e89a1205e41ecbd9be2f582dfea11 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3168c5611aa10d7537a35ee4780d7a7b0e9d4432","revisions":{"3168c5611aa10d7537a35ee4780d7a7b0e9d4432":{"kind":"REWORK","_number":1,"created":"2023-09-21 22:42:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/18/896118/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/18/896118/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/18/896118/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/18/896118/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/18/896118/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/18/896118/1"}}},"commit":{"parents":[{"commit":"bc2b54bc7526e5794bcecde7ec8292f6cb125057","subject":"Merge \"Fix TsigKeyring issues with dnspython 2.x\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc2b54bc7526e5794bcecde7ec8292f6cb125057"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-02-18 03:07:49.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-09-21 22:42:51.000000000","tz":0},"subject":"[CI] Reduce workers and disable cinder to save on memory","message":"[CI] Reduce workers and disable cinder to save on memory\n\nChange-Id: I677ac79047286fb92343d3d895c38cf1a1076556\n(cherry picked from commit fbfb1b0dd80bbcc2a6d1f09259972657d66fd998)\n(cherry picked from commit 6e7b7e898eeb2834188cce12f4d669163b28c3f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3168c5611aa10d7537a35ee4780d7a7b0e9d4432"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3168c5611aa10d7537a35ee4780d7a7b0e9d4432"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
