)]}'
{"id":"openstack%2Fdesignate~897361","triplet_id":"openstack%2Fdesignate~master~Idf7a0e3116991e1ac3f5443f0f133bb1601f163b","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-05 16:13:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf7a0e3116991e1ac3f5443f0f133bb1601f163b","subject":"Set a minimum max_request_body_size for tests","status":"ABANDONED","created":"2023-10-04 19:46:33.000000000","updated":"2023-10-05 16:13:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"283ee24bb267d5d85fdc826880b24def7e1f1fc2","_number":897361,"virtual_id_number":897361,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-04 20:46:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-04 20:46:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ceb03bf33ab64d04c277e68dee0902b7c3864a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-04 19:46:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21027af647ad1f9377683099c6bc0a44554bd291","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 20:46:50.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d6f26c208d14be788014322271bb7d7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/80c4193089034d9798ab9cd5f1dc502c : SUCCESS in 34m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a9f55e1c99e448adb6785addc7768f1d : SUCCESS in 49m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/956d43a4c3b04b75bc555f2ae7b7f4ec : SUCCESS in 43m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9e7708d437b641e89aaebb67bbc8f32a : SUCCESS in 34m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c707291d8774ed6bb382564250e1437 : SUCCESS in 33m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dc6ca53ab98645c9a91d927be868b331 : FAILURE in 55m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e669fee3137e4eb38e9516f138766f6e : FAILURE in 56m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d075afa588884aaba0845fdcb0aad732 : SUCCESS in 48m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/46030571f9cf4ae19ef3980e6b5d8ed0 : SUCCESS in 50m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c4abb1febe43a1b80f15bf69ab076d : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a770dbc86ff54bd08f0f4c606a8f3b8e : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66d9d11a943e4b8dbd36ff847e413bfa : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8881ebb4c5ba47bc9aa47e86248afaf7 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7893eec6c75425eb1d2c0574c6dc3d4 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7116a9f388be400aa6e2ccee46d8409e : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040fbf12111541f1b596e28dadde8902 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/982d0e6c4c4944ea8445f0addd311210 : SUCCESS in 44m 15s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e64370b566724d79a18f663a874a8484 : SUCCESS in 7m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6bf20bcfd4d1435bbb92ef2651888596 : SUCCESS in 34m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c8b6ff4fef8a4bffa0ecf103cbaa8022 : SUCCESS in 5m 50s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 89, column 3","accounts_in_message":[],"_revision_number":1},{"id":"283ee24bb267d5d85fdc826880b24def7e1f1fc2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-05 16:13:36.000000000","message":"Abandoned\n\nWe ended up fixing the test cases","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdc9f8357c51220117152f4fa96db3d779a6b00a","revisions":{"fdc9f8357c51220117152f4fa96db3d779a6b00a":{"kind":"REWORK","_number":1,"created":"2023-10-04 19:46:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/61/897361/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/897361/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/897361/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/897361/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/897361/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/897361/1"}}},"commit":{"parents":[{"commit":"d2d5e2bfc6a33a206920a70edfd334371de20dbb","subject":"Merge \"Add middleware to limit request body size to api-paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2d5e2bfc6a33a206920a70edfd334371de20dbb"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-04 19:44:21.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-04 19:46:31.000000000","tz":0},"subject":"Set a minimum max_request_body_size for tests","message":"Set a minimum max_request_body_size for tests\n\nSome tempest tests are exercising the maximum size of fields that are above\nthe oslo middleware max_request_body_size default size.\nThis patch sets the designate configuration to have a limit big enough for\nthose tests to run.\n\nChange-Id: Idf7a0e3116991e1ac3f5443f0f133bb1601f163b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdc9f8357c51220117152f4fa96db3d779a6b00a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdc9f8357c51220117152f4fa96db3d779a6b00a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
