)]}'
{"id":"openstack%2Fdesignate~898603","triplet_id":"openstack%2Fdesignate~master~Ife34b32a5e557d2bf893ddc570b0962622129f93","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-10-24 04:37:57.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-24 11:33:48.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-10-24 10:17:01.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ife34b32a5e557d2bf893ddc570b0962622129f93","subject":"Fix python shebang","status":"MERGED","created":"2023-10-17 15:49:41.000000000","updated":"2023-10-24 11:35:33.000000000","submitted":"2023-10-24 11:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"898603","meta_rev_id":"36b5d9629f8c81cef66d8229ccbb79e1bec6f1b6","_number":898603,"virtual_id_number":898603,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-24 11:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-10-24 04:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-24 10:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-24 10:17:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-17 17:27:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-18 05:59:27.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-10-18 10:35:50.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2023-10-24 04:37:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-10-24 10:17:01.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"391f3e19e3d003d1da2de9d6e71e0a4c864791a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 15:49:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bb225027970188bdabe40d53fc6fb8ab071c3f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 16:04:13.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9a233ca51c920cbfb975c0ab3e1304e44c27950a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 17:27:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3abbb67a87f4d2c983fdc92b68ea769\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c898a1e9d79a49e7b461d2ef1481a17e : SUCCESS in 36m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b2460122ad544a7cbab5e2ec515db0f9 : SUCCESS in 54m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a401d95f1c9f45a1b9e91e17ac125ffe : SUCCESS in 48m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d01997e0816d4df6a2f70e2689709191 : SUCCESS in 37m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/81eb1a11521241329e91120ffefe4f31 : SUCCESS in 51m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5c8f4386125f4ff0ae37499100c146a2 : SUCCESS in 1h 13m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b53138e336c54125b688334db8480a32 : SUCCESS in 1h 05m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3380f9d31cb54a7da9d3b124d8a70eb8 : SUCCESS in 53m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4e04b390448848cbb978c183844d888a : SUCCESS in 50m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e3ccf763d8744ecb190e9c02d114d28 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2cec64fc3b44c4966a8e98aed348e5 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa05af00329d4058a26613366626c8e9 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/930543da6aee409d9f530b717460c932 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7b5631dd6004e5c958735f920708a27 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ac8fcccc2e24938b75edc238903b442 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a40fa3e7c0c4771a2df85b99b5b7da6 : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/83e3d5f5ba2b4dedbd1ed2a3e7f0f7c0 : SUCCESS in 41m 25s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3121165af8694e8aa3352940620e9031 : SUCCESS in 6m 58s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b1a0df7bc2344c9a9bdd70d5b69c159f : SUCCESS in 49m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e54ce1e79855456b9ab6bb1c2013a50f : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02d2f2f95af79e1fd90dc01755ce619b7ff0dc50","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-18 05:59:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87bcfe9ccd3dffe44845620fcd4777b7ffeee961","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-18 10:35:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5591acaa9eb5f0683fea6025bb5ce44d9bca5e3a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-23 17:21:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a562f466bf9fe2b18901c80a05ef8e2680460e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-24 00:44:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"95e8d85e46a615d4a6e64e793189e6c954c4ba15","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-24 00:45:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e5eb60f3c14f4a2aa576202e8c1ae7053c3671b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 01:35:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07693df5db9a415d873504430f74bf12\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/af91e633b603424d8cdaa1828907fa7a : SUCCESS in 36m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bee9c1f795994667ac74218da16e2cb7 : SUCCESS in 38m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c7258b70d224e3aaa7c8967f5b2769e : SUCCESS in 34m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2d09bfdd376d443caf46f0d541e71cb0 : SUCCESS in 30m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/26cdc74831934d3b8b0072f0de57437b : SUCCESS in 31m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/61dfd7dc136d4b3da02e3212c2f25809 : SUCCESS in 49m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f91622b1a15447d98cd7c68987901b87 : SUCCESS in 47m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cceb29c2c09940609215c837355980ba : SUCCESS in 34m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/87d66d507f7c46f69a9b9ba089e5df70 : SUCCESS in 34m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df002deef346429293251a68e1499d8f : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a331b042b284640a5a2701b5ed54511 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/127f793ce9fc415982913b837f94c4b5 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4e869e4b3814a1aaefbc0f9ee6fee71 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0520afcd83a44168447298697922abe : SUCCESS in 6m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a254e94a4cb149468bef898c4c45eed0 : SUCCESS in 33m 16s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/114ad65c13c443ffb2ceb85ad50c2069 : SUCCESS in 7m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e04400af89b34884b52acb5543dd82c0 : SUCCESS in 32m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5699e9e4062b458cbc8949dda17a9a92 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d93cad0bef48e9b2aa470b7765c3afa09f35835","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-10-24 04:37:57.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13a46a55401b493376c4ed23de7edc365c2a4c80","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-24 10:17:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1dd35f73a7b40cb6c358c774896420ab7ed730a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 10:17:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"53212153191cc0badb2d233ad9b4452a9a32c851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 11:33:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dffcd39fd44f4c32bf3b24a72d16def6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5c6b3882989c4825becf8a0c82b6de3b : SUCCESS in 47m 10s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c958a1f6073e4453b41e421210bcb1e9 : SUCCESS in 45m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a2e58867bff44037a5ef21de3d730516 : SUCCESS in 49m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dba6637006a3457c8392091f2883a1c1 : SUCCESS in 1h 08m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2da922ad71be43c6af44fcdedeaf11a0 : SUCCESS in 47m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/33ab20507f5444758db8dcdd4e0ed755 : SUCCESS in 48m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc98ed3dc3d340dbb0d62f03676832da : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eea8c2793d4f4a86a0dcf12bd20acf79 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8456f25d409f4553a1da80cfdcd9125c : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cb3ad3d26a4f76a0084af134ca2958 : SUCCESS in 8m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ce2f3e5d89004ec8aae8312116c00ef0 : SUCCESS in 33m 10s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/54170325b46d48b79ad6174b7eb390aa : SUCCESS in 7m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a48194663b5f46ffb5ef97911d4a4a14 : SUCCESS in 49m 41s","accounts_in_message":[],"_revision_number":3},{"id":"2b2c60136327923fb8b1edf3c48af2f8f38a5331","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 11:33:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"36b5d9629f8c81cef66d8229ccbb79e1bec6f1b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 11:35:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b9a4120f602413a8b56503398ffd4b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff99831292084fa690353c6f7098ab9e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"24a58fce9d25918fb8e122f7829159ad2e59b9de","revisions":{"15dfe4175e0f6ed2158aa181a35063e924a74ff2":{"kind":"REWORK","_number":1,"created":"2023-10-17 15:49:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/898603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/898603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/898603/1"}}},"commit":{"parents":[{"commit":"d2d5e2bfc6a33a206920a70edfd334371de20dbb","subject":"Merge \"Add middleware to limit request body size to api-paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2d5e2bfc6a33a206920a70edfd334371de20dbb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:49:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:49:38.000000000","tz":540},"subject":"Fix python shebang","message":"Fix python shebang\n\nThe current shebang requires python 2 instead of python 3.\n\nChange-Id: Ife34b32a5e557d2bf893ddc570b0962622129f93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15dfe4175e0f6ed2158aa181a35063e924a74ff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15dfe4175e0f6ed2158aa181a35063e924a74ff2"}]},"branch":"refs/heads/master"},"39adf1f4ef63fd6358febb50106e26565869f0fa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-17 16:04:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/898603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/898603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/898603/2"}}},"commit":{"parents":[{"commit":"d2d5e2bfc6a33a206920a70edfd334371de20dbb","subject":"Merge \"Add middleware to limit request body size to api-paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d2d5e2bfc6a33a206920a70edfd334371de20dbb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:49:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 16:04:13.000000000","tz":0},"subject":"Fix python shebang","message":"Fix python shebang\n\nThe current shebang requires /usr/bin/python which is not available in\nUbuntu Jammy by default.\n\nChange-Id: Ife34b32a5e557d2bf893ddc570b0962622129f93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39adf1f4ef63fd6358febb50106e26565869f0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/39adf1f4ef63fd6358febb50106e26565869f0fa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"24a58fce9d25918fb8e122f7829159ad2e59b9de":{"kind":"REWORK","_number":3,"created":"2023-10-24 00:44:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/898603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/898603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/898603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/898603/3"}}},"commit":{"parents":[{"commit":"dc70b921408d06e826810d2efe5cf7df49fc119f","subject":"Remove legacy unused devstack code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc70b921408d06e826810d2efe5cf7df49fc119f"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-17 15:49:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-24 00:44:11.000000000","tz":540},"subject":"Fix python shebang","message":"Fix python shebang\n\nThe current shebang requires /usr/bin/python which is not available in\nUbuntu Jammy by default.\n\nChange-Id: Ife34b32a5e557d2bf893ddc570b0962622129f93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24a58fce9d25918fb8e122f7829159ad2e59b9de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/24a58fce9d25918fb8e122f7829159ad2e59b9de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
