)]}'
{"id":"openstack%2Fdesignate~899541","triplet_id":"openstack%2Fdesignate~master~I552373e99fc184c5da3193d0a44a08da6dcf3e51","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-10-31 10:49:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I552373e99fc184c5da3193d0a44a08da6dcf3e51","subject":"Removed deprecated mdns config","status":"MERGED","created":"2023-10-29 12:16:19.000000000","updated":"2023-10-31 11:53:52.000000000","submitted":"2023-10-31 11:49:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899541","meta_rev_id":"a316d38d8839a0a55b9015b91d76bca7d2fdb608","_number":899541,"virtual_id_number":899541,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 11:49:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-10-30 23:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-31 10:49:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-29 13:11:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-30 23:33:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e7c231f0f75689fc8863a446db030fd58ea400a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 12:16:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37dc74bc28f1ea2a72cb223ceed65f8b205198fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 12:21:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"65add034739a4df0dafc90c044d0eea743538d80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 13:11:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a17436be50054275838275f9da4811a8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9da4f94db93a4ecb929913d5c9690235 : SUCCESS in 33m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8818c40eeca045b8a3dda17af4864817 : SUCCESS in 39m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7864cac7a2f248d687f240828569d554 : SUCCESS in 33m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cd13bd0fb5644bd09248fef9124781f1 : SUCCESS in 30m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ba1b8422b5d94ab79cede4dd777a0413 : SUCCESS in 30m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6e6511b1670b48afa99ae05371d1adc5 : SUCCESS in 48m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2605a7d24eaf4bbd94c1b73331a6c4d3 : SUCCESS in 47m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8a32c56510ac4bbc91e4256d5ce4d98b : SUCCESS in 31m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c0937c66c4ae437491c0117d852e2918 : SUCCESS in 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4e7cc72ebe74743837de422f2e6d9f3 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833fd0361f574241b487b0701a708861 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c43f56d355724c0a97a0cd9704f61968 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed168613f39e40e4b8419947ec54f79f : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524d9fb1ba9d4e18ae77307b84b72975 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7eb142a1aee4f608712062b885b1e49 : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8995c8ac45414fe693a769eaecf307d8 : SUCCESS in 32m 38s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b04f6db4d8b04662b53da09dab650878 : SUCCESS in 6m 22s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0073764699244ff2800eecee68282452 : SUCCESS in 33m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5dd7485f25aa4191af2546c65585148b : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cd18bc92d516769f4ec1e6c22498151f1b6b6ba4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-30 23:33:07.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"492432659174a4f55ac18b3827619a28409a138a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-31 10:49:22.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac0128baa60f594acbbd7adaa638f33f99fcc395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 10:50:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba0e853ad7868a73d6f81017cb8a4659debcfb31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 11:49:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be162fbf99f84f65a7ec9609a4059fe4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2fa12d0d8b9f430eb0c57d7c680ea03d : SUCCESS in 36m 37s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/edf6e3b5af1c4219858c15198370b078 : SUCCESS in 39m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0dfc5f964ff74a06ac7a475ffb2cbd3c : SUCCESS in 38m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bac8897971ea43e9a2965daf5762be8d : SUCCESS in 57m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb5baddea9a741119c3c59ff9b6deeb5 : SUCCESS in 29m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/68c98f6d1aec452b9b59c2245fdb77ed : SUCCESS in 39m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb9eb63633394c9b95ad3eff0c16731b : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b807f8ec4f88491790c7000b7274414a : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be23f4477152478d84d38108ab3c8654 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6cab26f61d41e28dc0c05d2f650b1f : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ea4e84a26cf478d8d5c2b17ee143624 : SUCCESS in 5m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bc67c671f8924b84bdd1d06f4deccb3b : SUCCESS in 36m 30s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b158440bfdf84009afe8231446357cf4 : SUCCESS in 7m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/28b83dac16ed4b19acff64265de157d1 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":2},{"id":"c48ab7ac7779b760cdb3611a939e0c744bf0f71a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 11:49:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a316d38d8839a0a55b9015b91d76bca7d2fdb608","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 11:53:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3032b77e1784f498fb1808766589174\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aed5747ba776448b8c3d5122877e0a1f : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/945111e1988248ba90b437abb6ed0209 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"329e292c870f63fde1c46cedb7e7d22299ec5866","revisions":{"dc3b83564f344f8d2d2b4160abc74bd59cf190a8":{"kind":"REWORK","_number":1,"created":"2023-10-29 12:16:19.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/899541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/41/899541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/899541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/899541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/899541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/41/899541/1"}}},"commit":{"parents":[{"commit":"6354b5f79c4de234315b94deaba63907e889a80d","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6354b5f79c4de234315b94deaba63907e889a80d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 12:16:07.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 12:16:07.000000000","tz":-420},"subject":"Removed deprecated mdns config","message":"Removed deprecated mdns config\n\nChange-Id: I552373e99fc184c5da3193d0a44a08da6dcf3e51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc3b83564f344f8d2d2b4160abc74bd59cf190a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc3b83564f344f8d2d2b4160abc74bd59cf190a8"}]},"branch":"refs/heads/master"},"329e292c870f63fde1c46cedb7e7d22299ec5866":{"kind":"REWORK","_number":2,"created":"2023-10-29 12:21:34.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/41/899541/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/41/899541/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/899541/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/899541/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/41/899541/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/41/899541/2"}}},"commit":{"parents":[{"commit":"6354b5f79c4de234315b94deaba63907e889a80d","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6354b5f79c4de234315b94deaba63907e889a80d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 12:16:07.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 12:21:19.000000000","tz":-420},"subject":"Removed deprecated mdns config","message":"Removed deprecated mdns config\n\n- Additionally removed unused topic config.\n\nChange-Id: I552373e99fc184c5da3193d0a44a08da6dcf3e51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/329e292c870f63fde1c46cedb7e7d22299ec5866"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/329e292c870f63fde1c46cedb7e7d22299ec5866"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
