)]}'
{"id":"openstack%2Fdesignate~899544","triplet_id":"openstack%2Fdesignate~master~I1aa8c99f36f2373feee37a15376aee5f45ccc2d2","project":"openstack/designate","branch":"master","topic":"pyupgrade","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-02 09:25:01.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I1aa8c99f36f2373feee37a15376aee5f45ccc2d2","subject":"Improved dynect backend coverage","status":"MERGED","created":"2023-10-29 17:01:18.000000000","updated":"2023-11-02 10:36:32.000000000","submitted":"2023-11-02 10:35:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899544-pyupgrade","meta_rev_id":"d532a2eb3fbd8f33cca88da7cddcaa4add268829","_number":899544,"virtual_id_number":899544,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-02 10:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-01 23:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-02 09:25:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-29 18:03:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-01 15:34:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"dda75a134d7fa939a0e26082605ad2146de908a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 17:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da3dd62fce3db3d151d7bd72423a0553bca7e9f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 17:01:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fc5664e40654fbd6f80ac45976db191c605ae746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 18:03:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e946f41f7b4439caa4c05b0f425cba0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/395d8010ef26416dabd978384f412fec : SUCCESS in 39m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/95e47c738c8b49aa8a14b8c9d259c535 : SUCCESS in 45m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8297ea0acae413f8930558d69cc1f33 : SUCCESS in 40m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fee0cc7cc0c64fad84bbce9efc280b7f : SUCCESS in 36m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fdaa2f666a2f40078885c76416463819 : SUCCESS in 39m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/18da0f927dfc4ab49c24e13b026aed7a : SUCCESS in 1h 00m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/08e03cf0addd41e7a9b9ecf36ae67651 : SUCCESS in 59m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f0211b6546884f15925a6584c80f6fb6 : SUCCESS in 42m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d500c09f51034f2881dc9e49c6004630 : SUCCESS in 33m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e751cc7656534c76972bf1d4cdb9bf3d : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac07e1fae74044c9b04251893a4fed49 : FAILURE in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf2071cc5d940c1a63920b03e1aad0c : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db1500eb0613456fb60f5d4068684016 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa98c0292f074546a3024e3720eda493 : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/107e77bd7d334cbfad6440e393f10b73 : SUCCESS in 35m 20s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/00ab78e7390740aea7fe6acd135c12d0 : SUCCESS in 7m 12s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bc62fbdf68ee4f3cbabff309a7e84898 : FAILURE in 42m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/54f04536acba4b0ea6ea69b23536e4db : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8040833c49f16af215c4c56504d726919350a521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 18:08:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ba99661939ae1c3e820cc61d7141ec68ed41254d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 18:20:09.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9d0121398a19933c25850250acdf10ceb70b229e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 19:14:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859ee7cef9084a0891248e513303335f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/62f9506f26c648a69c8d74bbfec8c730 : SUCCESS in 34m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4d4b7f0a73494465af713bb5861a2e53 : SUCCESS in 40m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/208f23ef863a4ddf893e1163aab7ed1c : SUCCESS in 35m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/731cacdcc0024e318c7000f71f86aeea : SUCCESS in 32m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c9722643383345b6b0d0c614120a3b54 : SUCCESS in 31m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bba9e58a2bb7499ba96b321396dd4fe0 : SUCCESS in 52m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d9efc0f1f1c74b1d8b0ccf6688001569 : SUCCESS in 50m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ac0003a9c8df40d68a0958498119c04a : SUCCESS in 34m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bd8b575ae367498991c380d859839c03 : SUCCESS in 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bcee3edca4745bcadf74b962ae6a770 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a64095747df44aab9a2da3f7c09e2b92 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c256543232ee4de682669b9e716f7a6f : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b5a8606a9064984be0da144c8a1f93e : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50f0bafa71d4421b11b5afacb25bf5e : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e74a7b6f416b416b9365e1b1b24a6602 : SUCCESS in 32m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fe94a0c0fac548719fd0373ddba6d258 : SUCCESS in 5m 55s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ba56da9841764b01a5e2651c9d560f38 : SUCCESS in 31m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4185278739df423196c4a8945c2dd6bb : SUCCESS in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"205cc420e6b4c087bd133f283681a17b3bfc6dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 12:23:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"4cf8e02f2c7860dc994dba96b906795dcb580ada","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 13:22:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cffacdbb270b41d7b56f79fa8b9a3ed9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/be8b052ac63747778988b9ba6f852eba : SUCCESS in 52m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/300ab3d246df4f1eb2af565afd9e12d8 : SUCCESS in 43m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/99905c8621ef40c89edd1810b971e304 : SUCCESS in 37m 52s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/923ca1783bae44828420aa77377619ac : SUCCESS in 36m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fe619b55d9134cf9aacf0b495b3db402 : SUCCESS in 49m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d3d08fe4bf59412ab6c2d5e45f69c82d : SUCCESS in 57m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bc437b78713847a5ac2ccfa11b77a757 : SUCCESS in 53m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3a11b72f1d940a4b5bfce41001a3aca : SUCCESS in 36m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/290b4762ae90469dafe6aa902407adf6 : SUCCESS in 36m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57587ad7eaf4e4581352a1f5d891512 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39d444a642cd4524b1110b8f4922ab37 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f66579e6dd8467d8792b6f3a6bca425 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d62ecb1ae29d422d88e2c4be62dc4998 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6643e8d71a2d4a88ad23844e14abd1a1 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65ed493024ba47f399dea091f4148cf3 : SUCCESS in 36m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/610f8fc6f49d4313bed53b91e713c5ab : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/13cda48f69db48ae829e3ce43cab2c82 : SUCCESS in 47m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3f332486840f440b9a234898122c6d24 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8bad144cecfee50891fb55a67d82326a2a94b7f1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-01 15:34:01.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"84a9d6cee5b39f106b4887cb3c0dd8b10d0072e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 16:25:07.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"600d575a1895774800d4c7c144acc9cf0f1f2934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:41:25.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc991637a7ba453a83b207227979a89a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32e62356009e4b9f9253c3223bf27837 : SUCCESS in 47m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/71e341fa51394f959a73ce6fa0c63c22 : SUCCESS in 54m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/56a3b938a04c4859bfa9740ba72286fe : SUCCESS in 52m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/05f3f2d2271843e1b207964ee6099a1e : SUCCESS in 41m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/070d739deadc445dada6c77074a9e9cd : FAILURE in 31m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/992c1f40032448feb6112b5a6a36d298 : SUCCESS in 1h 06m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/870d4ab1cc3242c0825caa81d19f0365 : SUCCESS in 1h 08m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a5ba68a294384e46a0e90a499c9ff0ac : SUCCESS in 38m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec03be52a5f34d47b00b911d81d7fd2b : SUCCESS in 36m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/238bb398fbfc4399925d114efdf6a037 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55310f204d614c9aa1df44a07802c308 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efee43fc0bde44b3b5e10426a58662b3 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f6332d883c4477faeb1839365eba29e : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea5ee991fd5480591bced15cd0deb5b : SUCCESS in 7m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5d1e9d0e7f0d49f1ae24f4137589c4d5 : SUCCESS in 33m 33s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f4497c1ad81f4ab690982ff5d9610291 : SUCCESS in 7m 08s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/23a928fb000f4efbb0888d5ef61d777c : SUCCESS in 39m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3a62a738121843e4a220afa84f87cfb3 : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce98694d4863ed5f90ecb723f07e6723d8020f77","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 17:42:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"59d2d08812e640b85bff2dca6f942d31b43dcf48","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 18:02:08.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"516c07774e9603bb8afb3604dbdc5321acfd5d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:01:47.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2633b56b29e74617b7933093ca3109af\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fb2a3a975db4d87aa9879336c993d15 : SUCCESS in 45m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1ead0de67f0d4caeb26a55c84a2fb59a : SUCCESS in 41m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b327ab04dafb4f6896660fe0b3af29bf : SUCCESS in 55m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/46d7aa5786de41bc9dbf0e4d8e26c8dd : SUCCESS in 44m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b6f1c24a8942442493a1e78981649d3c : SUCCESS in 36m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c5aa4949c50341e6bc9f65fb5a1bc88e : SUCCESS in 1h 10m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7d7988dfac0488d9811082a8d48e765 : SUCCESS in 1h 13m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/16f40e7b855848e18e6358aed4874a69 : FAILURE in 46m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e06088973d404b418995febd4b6bc37e : SUCCESS in 51m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e42d74e5681493b8cf5c636aeecbcfa : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c31e501fcd24ec8a4e1b60ca588f99b : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50ab899987d4f028c4cd926d531748f : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5d16af4f3b643219a8fb05eb924b454 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f3a01b3564452380e21ce8b37e823b : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f6ca315ad1a147fe8f118e5b01e5216e : SUCCESS in 38m 56s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/03ac8de2dff44462b95e7ca1a34b42ee : SUCCESS in 6m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a071343b79774df992ce50b714a15ce7 : SUCCESS in 34m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ed86a612d0f448b78d5bfcd9de9491cc : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9e7a432983ef4762cb57844504aa093f533b3f57","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 21:00:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"993eb9f0878331f73c9a663cba1069ab631afce5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 22:25:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aa59d63a7dc49b98fca6d572d45c934\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eed9177eeb4c4b18b7950feebd6a0b95 : SUCCESS in 50m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0435761e4f3c4709ac0460344b7796dd : SUCCESS in 1h 02m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/49edee88c069441f991ae158fdf4e123 : SUCCESS in 57m 32s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5b0bab07ac70407a9e9a0c8d55cc131e : SUCCESS in 48m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3dd2e72b07f24c91aeeb287d72c1444e : SUCCESS in 49m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/46cb6c9e3cb242ecbb2835fb93795dcc : SUCCESS in 1h 19m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3e162974af864210823faa0a8d9efc03 : SUCCESS in 1h 10m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7be4bb15f03041c1a0cfecb28bd073a8 : SUCCESS in 39m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0123028a176048809c2d3ec09a758b0e : SUCCESS in 35m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/998d72930bee48aab478428486cff5c0 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6d436d4d9148a5a7a506d37c6e8673 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc94a8b41f0942358954af997f3dadde : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c14c2258252747b095a4514a32f6d967 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/274e5b84c2f44dd7864d95f362011e9e : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b8bbc6ec40994e4db1ee780b69a0d994 : SUCCESS in 33m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5f280e37f5c346b6aa8e8ef423dd96db : SUCCESS in 6m 20s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7a548520c53944f8a6aee08b022290f8 : SUCCESS in 32m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c3f1854ddbe54022b655caa0868f6e4d : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b87bc0cf4badb55ffd9eb0c8117a98f12dbb9347","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 22:26:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"f53b4ac3e9b5899afc171dfc80ec3c269c78fc69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 22:29:55.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/899108 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"639ce2c01bf599114890baefff41528c25f90c3f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 22:58:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3550e969d6342d92d18ae399d3e62dceeacff7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 23:01:49.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"97cf0396ea2dd26b18fa6d1f7529cfe37fef6108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 00:17:48.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9568035700de4e64b03fec5160e2a4b1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7b712be68fb4486c852cfcdbdc3a2006 : SUCCESS in 45m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b7acb2e0e5804f119aa535420d18ca5b : SUCCESS in 36m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cea1c6841c45444a814aaf998478be3d : SUCCESS in 40m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7d6daba164a54cba970444b6b3e7baed : SUCCESS in 44m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2f463cb71b2244488be1d9d8537a69c7 : SUCCESS in 54m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bcf4929fd5e642d9a79ff2821e61d956 : SUCCESS in 1h 08m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9bbba72521724bc298df4500f7454639 : SUCCESS in 51m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f98507d5030c4a03ac0bade6cf9e11df : SUCCESS in 30m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/97b6d36abefb42b894dd04f350ed5f71 : SUCCESS in 50m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/241f6852fe874d01b929e3afaf7125a1 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2fd4128b2f54936b6333a5a1fefcff5 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16aecae614b64aee8db6652f21e95b2a : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb45f155312a43b282e8e3e6693a5df1 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5677c7abe6dd471abc0ee6cab2e83e81 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65d351bb11a14a2a82a5f9546a2bfc5e : SUCCESS in 37m 24s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/350111f0e57e46c1805b24d1e4dc7fea : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/27f3e294fcb543eb9b58201da2b3e5e4 : SUCCESS in 39m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2a276201070d4330bc52e8a82a9b0c03 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d315ab93194f2b7b30cc275340855ac6287edd85","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-02 09:25:01.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c5064e1f656561667c5995bd8d96e7a3bfb6aa2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:33:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a36190331493c599d5abb7962de805b772ea089d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:35:13.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c2514e99491482aa36a3d982a96fcd0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e2fa082dd1149b7a84f6985150557e9 : SUCCESS in 38m 56s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c95eacc6949a41388ab4e27ce18274cc : SUCCESS in 40m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e8b8074c516347a896a191b7c921056b : SUCCESS in 41m 03s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ec90187ea6514bc8a0018b8d3c371b7d : SUCCESS in 1h 00m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5a2a8570deb34272a25e146b4e9cf5c0 : SUCCESS in 39m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f16f71810e374121bc62a6e662f07e66 : SUCCESS in 38m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60bc0b0409ab4ff1947f86fc57ff5547 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cba0345c4f82487fbb8279cedfa67e26 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2075ad6f91d4e31855f9ebe2fc899c0 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abc9b39c241493abb45675b2c21a458 : SUCCESS in 8m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d943ac4d92145b2bf6e904533921a4c : SUCCESS in 36m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cc076c6f475848938a18118f9406428a : SUCCESS in 7m 27s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ca831fb30cf5497c8f96541bbf8591eb : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":7},{"id":"c8c9778a81108ff873df2a5c8229b25b349bc67c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:35:13.000000000","message":"Change has been successfully merged\n\n5 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":7},{"id":"d532a2eb3fbd8f33cca88da7cddcaa4add268829","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:36:32.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748ba0bb5e684276abfcef6a856f1019\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c038391e34f4a4c95754a744935aa21 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"0b9635ddeceeedd05ed109cc36bce125b342cb41","revisions":{"ee6de4ec8d1a95bd2252cd2361d557d906846e5b":{"kind":"REWORK","_number":1,"created":"2023-10-29 17:01:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/1"}}},"commit":{"parents":[{"commit":"6354b5f79c4de234315b94deaba63907e889a80d","subject":"Merge \"Fix python shebang\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6354b5f79c4de234315b94deaba63907e889a80d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"subject":"Improved dyntect backend coverage","message":"Improved dyntect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee6de4ec8d1a95bd2252cd2361d557d906846e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ee6de4ec8d1a95bd2252cd2361d557d906846e5b"}]},"branch":"refs/heads/master"},"329e59606045fb4f51536492d95df38fa9bd4175":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-29 17:01:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/2"}}},"commit":{"parents":[{"commit":"4b776b04729d3d5660ec88f116c3edb3c8159f52","subject":"Increase minimum test coverage to 92%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b776b04729d3d5660ec88f116c3edb3c8159f52"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:01:38.000000000","tz":0},"subject":"Improved dyntect backend coverage","message":"Improved dyntect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/329e59606045fb4f51536492d95df38fa9bd4175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/329e59606045fb4f51536492d95df38fa9bd4175"}]},"branch":"refs/heads/master","description":"Rebase"},"09aadf2f4e1727e8da360b0bd1e7500412e9130a":{"kind":"REWORK","_number":3,"created":"2023-10-29 18:08:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/3"}}},"commit":{"parents":[{"commit":"4b776b04729d3d5660ec88f116c3edb3c8159f52","subject":"Increase minimum test coverage to 92%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b776b04729d3d5660ec88f116c3edb3c8159f52"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 18:08:28.000000000","tz":-420},"subject":"Improved dyntect backend coverage","message":"Improved dyntect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09aadf2f4e1727e8da360b0bd1e7500412e9130a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/09aadf2f4e1727e8da360b0bd1e7500412e9130a"}]},"branch":"refs/heads/master"},"5cb0433eb982ee9d839d65a2ead0c2bbc1b9efcc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-10-29 18:20:09.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/4"}}},"commit":{"parents":[{"commit":"4b776b04729d3d5660ec88f116c3edb3c8159f52","subject":"Increase minimum test coverage to 92%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b776b04729d3d5660ec88f116c3edb3c8159f52"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 18:20:09.000000000","tz":0},"subject":"Improved dynect backend coverage","message":"Improved dynect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5cb0433eb982ee9d839d65a2ead0c2bbc1b9efcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5cb0433eb982ee9d839d65a2ead0c2bbc1b9efcc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"600bc0523110faffa798cd81aa08b1eb9c6eff5a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-11-01 12:23:23.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/5"}}},"commit":{"parents":[{"commit":"580ad1f49fd001b175d3c4878f92568bda8a677d","subject":"Increase minimum test coverage to 92%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/580ad1f49fd001b175d3c4878f92568bda8a677d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-01 12:23:23.000000000","tz":0},"subject":"Improved dynect backend coverage","message":"Improved dynect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/600bc0523110faffa798cd81aa08b1eb9c6eff5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/600bc0523110faffa798cd81aa08b1eb9c6eff5a"}]},"branch":"refs/heads/master","description":"Rebase"},"85430cc50385b4af1a3d4e6854b2edb77c09741c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-01 16:25:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/6"}}},"commit":{"parents":[{"commit":"a3e11127891d33ce7d6162a232b312748676302c","subject":"Increase minimum test coverage to 92%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3e11127891d33ce7d6162a232b312748676302c"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-01 16:25:07.000000000","tz":0},"subject":"Improved dynect backend coverage","message":"Improved dynect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85430cc50385b4af1a3d4e6854b2edb77c09741c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85430cc50385b4af1a3d4e6854b2edb77c09741c"}]},"branch":"refs/heads/master","description":"Rebase"},"0b9635ddeceeedd05ed109cc36bce125b342cb41":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-01 23:01:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/44/899544/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/44/899544/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/44/899544/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/44/899544/7"}}},"commit":{"parents":[{"commit":"0f7e2c70b98e0ff912e32fad81fa87c2e25c8e6e","subject":"Increase minimum test coverage to 92%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f7e2c70b98e0ff912e32fad81fa87c2e25c8e6e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:00:52.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-01 23:00:54.000000000","tz":-420},"subject":"Improved dynect backend coverage","message":"Improved dynect backend coverage\n\nChange-Id: I1aa8c99f36f2373feee37a15376aee5f45ccc2d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b9635ddeceeedd05ed109cc36bce125b342cb41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b9635ddeceeedd05ed109cc36bce125b342cb41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
