)]}'
{"id":"openstack%2Fdesignate~899546","triplet_id":"openstack%2Fdesignate~master~If04a2673c624390c59cb6fabc017b06fe52edc0c","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-01 16:24:51.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"If04a2673c624390c59cb6fabc017b06fe52edc0c","subject":"Improve pdns coverage","status":"MERGED","created":"2023-10-29 17:32:25.000000000","updated":"2023-11-01 17:47:59.000000000","submitted":"2023-11-01 17:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899546","meta_rev_id":"90b88512c8f95c4f0a416b7a5d16b091ebebc451","_number":899546,"virtual_id_number":899546,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-01 17:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-01 12:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-01 16:24:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-29 18:31:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-30 23:34:31.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a800f7667e5213d166b5a8fac4e6bd5746b8c3b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-10-29 17:32:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d034a29c5709c09d9fcaa1b8ef0c2e0c7d0f0f51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 18:31:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93fc2da5d0b045e5ad6b2732f81ae653\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b9779733bab644bca1d60dd69274fd41 : SUCCESS in 41m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/94bcb139b6654c1183304be1ac51cd69 : SUCCESS in 44m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1027f6efb6a14058867c50e8c76719eb : SUCCESS in 42m 51s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f23d91452b2741199e576c4cc4cc451f : SUCCESS in 36m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/647b720ac1f84468925fdefa94ae2b5e : SUCCESS in 35m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/747ebb73554f45d1ba3b48eb66d0c106 : SUCCESS in 56m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1dd5efc84c6e4a55a402fba009e9ab9d : SUCCESS in 57m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/609ecede0b84459fa6f5b9f854359a7b : SUCCESS in 38m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9184a140fb724e5486d851fc71d8ddd3 : SUCCESS in 36m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4e4c434cb9a4afb9c4a709a6015fbd7 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace4fec550e94e7f9a06fc68841ad144 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baaf40a2438440068316f202a859394f : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ecfb13d71ec4a49860d9711e0087bd3 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf27e2705cc94fce9d1ab0c135e69452 : SUCCESS in 5m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fdbb1ecf9e034d95bc4b2b6d0d00c9f4 : SUCCESS in 33m 31s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/85e3731f33d942378b55f1a698d23aa4 : SUCCESS in 8m 24s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c7514debf7b44209be9a4a13b2b15238 : SUCCESS in 31m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7a7e74cc08db4109800cbf669f50065f : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57bf55b276e2f537ae1185478020f914231b1479","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-30 23:34:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ed644bae103c7b1856c124a260646de4e1c8b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 12:22:50.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb1de17463eabb5f580b961188e93747110af490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 13:39:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2def564465c64fbea2e000a4571e60fc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8c2979ca4ef414991e4bedb6bccb619 : SUCCESS in 33m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4f899193dd764d4b8b3947dba49f1d15 : SUCCESS in 41m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d9901aacaa4436c9c79398f4a800628 : SUCCESS in 1h 00m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/324279436d9b44bf9717fd5f2932deee : SUCCESS in 34m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8975515b09ed4d66a3eaf8f934cebdf1 : SUCCESS in 49m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/06407c35d4b444e192cd5c3a702f3937 : SUCCESS in 1h 09m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8209efeef1b04b2cad9640fa538f96ba : SUCCESS in 52m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7683245e38844ee4ae391874f0d3ac1c : SUCCESS in 47m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4188707b39514a76b998f120e8feb47a : SUCCESS in 43m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/212edc4e7bd94f5db05042b0ec045f72 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5c0dd6ddd5462ba0ed2bfb9d05ba34 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9314a769237e41b29d91a67c3719e3a3 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af1e555d1ea94537ac355f37841a207d : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ac6f13b1f24111934d6b2c05a1b83b : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3c9be214fb774078ab4fa92990d62b4d : SUCCESS in 46m 19s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/73c7b296a6d748a39063fb75c3403748 : SUCCESS in 7m 04s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/13a5c6faf9994c14abdff9a14f6e3cb9 : SUCCESS in 40m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f70de7741d9b479cb49650c65bebe5a1 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79e71a015c606ac463ef3dd28d6c4f2cf5a4198e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-01 16:24:51.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d97b49fae65146db1e3bc7beaae9e97c9d9bc0dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 16:25:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6767ebb6600452edff9a456178c99bc531bf853","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:46:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a82852d243743de9240827af0a7014f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eb706c9e500343f8821a9338f47c66ce : SUCCESS in 39m 04s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/81be391bf9be4ebea20982890b751364 : SUCCESS in 49m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5b9a08f523074c2c9cf6a29bc620edee : SUCCESS in 43m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f71b0fdb074419ab4c36e1df9d7477e : SUCCESS in 1h 13m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a59d4f27f3c949f9901617a25a49683a : SUCCESS in 47m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5694938cc4aa4e4a8fd7b27f88c62606 : SUCCESS in 45m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31f47c529a34457db7202ec553b27120 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9050d46c8a8145858f79aa52fab0a38f : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d9ac312897e474684bd7802550257bc : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b56c5564fbb462682b4031a69a5e9be : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/64f9176af34642919ee04a8f400ae7c8 : SUCCESS in 34m 02s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6b2fb434c9ba43b69238e62eddd05aca : SUCCESS in 6m 49s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/29d4d39076ee4ff2be82a2bc79607ca8 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":2},{"id":"81c1eed2d7ef62c839f80cf680562e0d2ebe893a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:46:53.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"90b88512c8f95c4f0a416b7a5d16b091ebebc451","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:47:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6152d7f3a0d40f6a687e196d6f785d9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ad6c5b16bf742438c9371629448d76a : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"83985fea341d644be51b6e4b89016039bfcd1cc9","revisions":{"2e06870fc912a5af1c0fbeb71233f175ad1feec0":{"kind":"REWORK","_number":1,"created":"2023-10-29 17:32:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/899546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/899546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/899546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/899546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/899546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/899546/1"}}},"commit":{"parents":[{"commit":"bc95f1e62fadd32408c9977d9c3e04552b1e3fe0","subject":"Improve bind coverage and fixed minor bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bc95f1e62fadd32408c9977d9c3e04552b1e3fe0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:32:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:32:09.000000000","tz":-420},"subject":"Improve pdns coverage","message":"Improve pdns coverage\n\nChange-Id: If04a2673c624390c59cb6fabc017b06fe52edc0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e06870fc912a5af1c0fbeb71233f175ad1feec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2e06870fc912a5af1c0fbeb71233f175ad1feec0"}]},"branch":"refs/heads/master"},"83985fea341d644be51b6e4b89016039bfcd1cc9":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-01 12:22:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/46/899546/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/46/899546/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/899546/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/899546/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/46/899546/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/46/899546/2"}}},"commit":{"parents":[{"commit":"1b6546e137aae3289fe5b625715ae29489cd56ba","subject":"Improve bind coverage and fixed minor bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b6546e137aae3289fe5b625715ae29489cd56ba"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-10-29 17:32:09.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-01 12:22:50.000000000","tz":0},"subject":"Improve pdns coverage","message":"Improve pdns coverage\n\nChange-Id: If04a2673c624390c59cb6fabc017b06fe52edc0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83985fea341d644be51b6e4b89016039bfcd1cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/83985fea341d644be51b6e4b89016039bfcd1cc9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
