)]}'
{"id":"openstack%2Fdesignate~900103","triplet_id":"openstack%2Fdesignate~master~Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-08 10:22:17.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4","subject":"Improve object fields coverage and fixed error handling","status":"MERGED","created":"2023-11-04 22:36:21.000000000","updated":"2023-11-08 13:04:06.000000000","submitted":"2023-11-08 13:02:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900103","meta_rev_id":"876d76401d1c62a5599e83b1a963bd1ddbd1968f","_number":900103,"virtual_id_number":900103,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 13:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-08 00:57:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-08 09:06:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-04 23:57:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 18:47:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"81198d5ad66da975e54ec4a4d206ea53a0ece01e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-04 22:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0767b0f13546c558bd47fe74ac5f977d18db2281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-04 22:38:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbe9693ea40eff7757e5ca76aa0f5f620e072ecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-04 23:57:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58de3e74dcde41f5b682c589b09765fd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/37c130b758d24225af2182ac36f83478 : SUCCESS in 46m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/006a49e4c9c744d5aa8dec0cbd3bfcf1 : SUCCESS in 1h 00m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bf98172f7b443f0a87c786b20e5312b : SUCCESS in 1h 03m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0ea998ae0dd44982bf77435c19e77533 : SUCCESS in 48m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1cc25915b70b44589e700edda4936592 : SUCCESS in 45m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9dce592ad9af4df3b7c2bb15f175369a : SUCCESS in 1h 09m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d51338c7a934f9f871c10486d809ab8 : SUCCESS in 1h 13m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a1a7c6d42efe4204a7595649b3e32908 : SUCCESS in 46m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3aae5e2eb7be4cac899d0a0dba259205 : SUCCESS in 47m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98efd273b8354c8a9aeb36fc5bd6a607 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ef1e21e47145fb922322140dbb7a79 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a56db5c4197940979047403842cc0c57 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96944edd04404c0cb5cba002a2158ebb : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45701e8e62d4153842bcd4d89f17f39 : SUCCESS in 6m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b699bf86c99042ffa0563b7fc79e52a0 : SUCCESS in 32m 00s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fb46b506a9384ce490ec2f917d9aae8b : SUCCESS in 5m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3356444f43ab4990b854a51382206e88 : SUCCESS in 34m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/82760fab9e664a26a7e527837d0ae91e : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bae27b2943c61ef7b713d5bd23b75128b0d1918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-05 00:29:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1f42ca38a771d1f2c765f344e950888cc595be20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-05 00:45:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a8f74f6898736a47520367763768e618c0b095e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-05 00:49:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"05ff522b05464a6b0e043e854953f6be01fe2d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-05 00:50:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5b77cc65c78442437d6c6eb2b5a40e9e87b427ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-05 01:59:04.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7105c2c8ef4702b4e68a85407d1cf6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6ba400b869c04b3a83e6c8fa00bb8ccd : SUCCESS in 37m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c103f57d3e52463f9851362f6ad2ab4b : SUCCESS in 47m 09s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b99e52ca02f1403585919e552903c802 : SUCCESS in 37m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/802e4497ce9948808f3fd08b9dd9d214 : SUCCESS in 51m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dbb91846daaa43bbb453777a2480c115 : SUCCESS in 47m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d48fa98b97b04ca0833c838c4ff15174 : SUCCESS in 49m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dcf1788520f541a496ffd3e1d66d0b7e : SUCCESS in 55m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f646ff329c845fb8ede62b53af37502 : SUCCESS in 32m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c56172ab62da47ea9430688d96d1b369 : SUCCESS in 46m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dea64f2831ec4c8baf6e07279671c5b0 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34b84ad25c9441ab9ab73c73cde5569 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5545d1e593a7495fa695271ad09b56bf : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6414c8f8cbdd4fb69937bf4bd6fea49e : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a72d33a6c4c47e7af91e15ae9e72eed : SUCCESS in 8m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd5d95db54cc4aa9ababcc87ae049104 : SUCCESS in 41m 00s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/835a2ae61eed461da020fb35f8987863 : SUCCESS in 5m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/baaf2858015740a7ac16ce60279b0b3f : SUCCESS in 49m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/516da6c7a8b140f79c5540867b553e3e : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1710c6d9f61f0342c8dfb02cc14e0ec02490ff5d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-05 22:14:10.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"06dc2c55036ba8b606f6963305d479ac1006a9c5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-06 18:47:30.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8c76d471a27646874683fc0d214b880e5ee539c6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-06 19:24:04.000000000","message":"Patch Set 6: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8ab61e3823eae0bc6a94d319d6232eb6fc7c01c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 19:24:46.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"23878834441ba2fda252bd34dff48c81fd431a4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 19:50:58.000000000","message":"Patch Set 6: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/900102, https://review.opendev.org/900101 are needed.","accounts_in_message":[],"_revision_number":6},{"id":"7252cb801718601f0a1bc40df04a8742f0147737","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-06 21:01:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"280e962b557c1fab01817dbb2afff63d92bb2538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 22:15:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66889a7be9441f0af46bb91110f9550\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae9d5dfa8b93423a86990f21637cb4b2 : SUCCESS in 31m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3e1b7ba469bf498ab08287d38fa72265 : SUCCESS in 38m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/04e81c49ce23464fa27a19b8aebb7710 : SUCCESS in 39m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/993a5880e5304bd888c3aca1bcf46305 : SUCCESS in 46m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/adeabc6c8fbd459ea104281a8c936ab2 : SUCCESS in 28m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba513c233c5c4032a3dcfce402ae1e27 : SUCCESS in 50m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e7e4d9830e4e410cbf7c628ad48715e8 : SUCCESS in 1h 07m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0074bb0b536463daf0d18e65448adc2 : SUCCESS in 31m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f5ea25df25d0416887ec8afb2bb69fed : SUCCESS in 38m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77ec4d921c6f4e5dba9045b3f76b3754 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61a8786294ab4584bf32fc0a5cb7de49 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70b9b1d5d0fc4f45b90b50c96a44da9f : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cd092aea44344df8af9968ee4a2fb775 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da1858c91be54ab28db3ecbb74e21af5 : SUCCESS in 7m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2524fe370a54e8ab3e488af03ea402a : SUCCESS in 42m 15s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/842c63cff8d8409ca95b41817833941e : SUCCESS in 6m 00s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/73ba29a990314d48a19989d82c5a0323 : SUCCESS in 34m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7f8af3ffe93b4b5a97b9caf7776aba4d : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"90368d548d45e51d05799ee4ece89cd0985bb66b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 22:16:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"0ad98b39ab42e45731bdcd3ea7ef2ecf6c2757f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 23:10:04.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5250693424f247f2a854d2178bf04a11\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad2c16563e694b0fa283f8e9664fb017 : CANCELED in 25m 48s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ecc2e2039380434fbfeb07be151b01f3 : CANCELED in 32m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ea3eab6217fd4e6591918348c1971d1c : FAILURE in 27m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/347ef812edc041bfa8e95f2047a467e7 : CANCELED in 29m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2087b5e2dd594ca88d3a81860da2f7f1 : CANCELED in 27m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/62afa80543e24527836c3796e0548a93 : CANCELED in 27m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb209001cc3f49f39495e7f1539bcc75 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d533eba76dbb4b82a6712b4d78c01152 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97e60101b47a42b2a196c9d2dfa92799 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d70611bafea4c24a7ff07c29c3ea6bf : SUCCESS in 7m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e9c2ebfd82904ae99f856d8b2ee5b4f9 : CANCELED in 32m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1a92745235e84feab98b9dc8e0983b0c : SUCCESS in 6m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/53fd073eb8664dc1a37e9459805f64be : CANCELED in 20m 00s","accounts_in_message":[],"_revision_number":6},{"id":"13e7bf354334ede23bd76f9a2117fcf67b4f834a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-06 23:13:34.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"47ead03b71ef138fe568f77c7e605787c5ba1ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 00:35:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fd5bf9e646b49a5b8a9956591873b50\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8776dbe31c04e7193ac977f277faf5c : SUCCESS in 39m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/01fcec466d0a47e2a0aa6156a9e54d79 : SUCCESS in 46m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6114bb4061764c15a513d41ae0e917ae : SUCCESS in 44m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3f91e29962fe470c9b17860994c12ee9 : SUCCESS in 48m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b71873aaedbb436b9383b0b9ddaae3f2 : SUCCESS in 46m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3a173d89063445c28f0696b1048886fc : SUCCESS in 1h 06m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1af0d64c8ff345149a193e0d34a750e6 : SUCCESS in 1h 15m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e9b2ee7ee94f4a2ba0e18b082fe9ad5c : SUCCESS in 47m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d4d743a795854f3a9e15e544a96307c3 : SUCCESS in 1h 00m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/066941b368aa464c9a33f5a8d9fe12dc : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb293ba03bf4a4e9fb3ad8c99668ae9 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/810c8415748640e68a798d844f260594 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1aa5908b512d4b47b9ededda0eb2bda3 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf9fe4932e64ec39662e87e617dbe27 : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/999908c762b74c5cbb1719e5f90526c4 : SUCCESS in 32m 53s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eef71801bdb74fb5b0d3e5500bd3f0e7 : FAILURE in 4m 31s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ad5b9b85e7ed48ae9259cd9687951f8b : SUCCESS in 37m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6a42179b6a0f414eb0392321ea38f10b : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ec22839891ff6016778a89f702d3604c9f2c87b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 00:57:43.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c538adf2d3eaf4c1daf1104368f97ea8764a4da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 02:01:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d812ab1fec04b86911efc9d96d95c27\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c8bcf3ad22144cc190f5a40d40737922 : SUCCESS in 50m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d898adc466884e4ea4425be191b97d0c : SUCCESS in 58m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffc266a0d6ad409d80d86335e065a8e8 : SUCCESS in 37m 43s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/870b900def0b4c73b6cefe060d522a45 : SUCCESS in 30m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e6e20a5d148d4ee1a8184e575c795dcd : SUCCESS in 35m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/13caca2432584945844ffbb9e97256a2 : SUCCESS in 52m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/053a05ccceaf40dfa6494c0832d81954 : SUCCESS in 50m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0eb20f0d8c634ac28eae97ce98247742 : SUCCESS in 35m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a28efb051ec64a5ebfbfad09eefecaf6 : SUCCESS in 51m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/553b3652cd8f44ae84eb9a513a859c4e : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b84d690d94429aa89248d06eb9c804 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a61fd888df6f464a9f81f6477ffe8225 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ce60e4504574121850b82f7b85ff550 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a416b8658011403eab5d926a49926840 : SUCCESS in 7m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7027775695a5476f80083f6cc7d80bc8 : SUCCESS in 41m 26s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b16f97b45948466eb4f12b82623535c7 : FAILURE in 5m 54s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b1256748a3404be18149f7c832d8e7b4 : SUCCESS in 51m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/768371b48b6443139c80ea460d97eafd : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d98c038ce0dc4355610847f994ecec3532a20ed2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 09:06:03.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"c9b30d529cf069cc24e85fa75cd00975fb1f8075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 09:06:39.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"c282279dc1e50bfe5b028009cdaa4e73be0d2c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 09:54:59.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7aa857d23a3f4ae8922222b971457556\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/94b725387d6d4aa4a43ba50e1682d4fb : SUCCESS in 37m 28s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/141ac3bd54354985bb4b647b8a19060e : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/915a7a2f42904e979acfbfd6ce878ac5 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/192e2909f0e04e78ba24af01baf43ec2 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8be801a02eb54db8bf85d13f6f2e7d05 : FAILURE in 33m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4472ecb976ff4495b3ce6b90297c0339 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/968b419fcd9d446bba507ceda1475a65 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/987685ebfa0d4037897619d102256b1c : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db39ba7b85144bc5883f76c4ffd05e41 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8ebbab5629242db8ccb81b09c786895 : SUCCESS in 5m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5dc9fb57bf2d46849bae3ff45f4b688f : SUCCESS in 44m 17s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/add1b0eaed85461096e5badf0609a5ec : FAILURE in 5m 39s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a1202d1af3ed4af0a33b077bfaed3e09 : CANCELED","accounts_in_message":[],"_revision_number":7},{"id":"ae301b20ca372d0077dbf3809b2f89171073f5cd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 10:22:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"49d09a0a3b08a4232a25b042cd00f34f0fb15bed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:33:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c67cf946dca4e25af5c0d6d4304da4f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6acdfd1358be4595bc81ce04e6f6af8c : SUCCESS in 45m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0022748d327a4c8a8cbdb4b69b283653 : SUCCESS in 1h 00m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/01458338b19e47a6a5bb013fbf28627c : SUCCESS in 1h 01m 21s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ee7b1da1e3a34a8d90c6b21d688ec7b0 : SUCCESS in 48m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/164e05d6b24a46daa9ce46d0d873ff6a : SUCCESS in 33m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bc2d9fb5496248cb99cc19bf44259091 : SUCCESS in 56m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9a7ead251682496c869da623d59e6c2d : SUCCESS in 1h 05m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a5ef3b16f46148a9aca9315d3927adb6 : SUCCESS in 34m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3d696b54a79c463086f287be663d501d : SUCCESS in 50m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02cb78efc7434f6183a797e040c93e36 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/335eda00b7504ac49eff6423aa85bf68 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/324c925854f24cacb04bbfc52398e258 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa051b9cf9c54fb2893a206cd6e35723 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9847ba0c7304dbcb42c48983f53d2a7 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4a8438a850841f092f947766fc1e1e7 : SUCCESS in 39m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e6994f73ac7c4602ba6fa0dc18b897bd : FAILURE in 5m 20s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fdd997ebb27645a586ba08c231bc7958 : SUCCESS in 49m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3ba7e81cab23457da10e62fd1860f317 : SUCCESS in 4m 49s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0ead673bea166c87c34254c3438b1519cf4e2903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:34:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f160417a4647f49631ee9be47a20562e0b806ce0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:02:59.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a4ba9e491c64c888241a4733377c741\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/710bf7ca53d24a8fb9c4f9d3a5bb57c0 : SUCCESS in 53m 46s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f5c21a46d0f347fea4af3b1377a9c391 : SUCCESS in 37m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8da19aa609e849ea8beb42709b1d7ff6 : SUCCESS in 39m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b4d8713e5efa4b01aa602041054adc9c : SUCCESS in 1h 20m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e0136d59ca2c4322a21ad44d24312b87 : SUCCESS in 50m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7b43631f716544b4b4f516f2fde1ff95 : SUCCESS in 37m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da185a29e8b942949009dc5629b9aeea : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e42a6e38552c4b9083995a4ca6e2bf89 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/32402c69ee544250ab1363d4d0caf891 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1529d8558b304ce4a65a0113c319486b : SUCCESS in 8m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/36b48b7b8a374061856aeefa3dc75429 : SUCCESS in 37m 16s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1306a620f4e64fe58a7486d54fed112c : FAILURE in 5m 36s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b32eefe8b7a34217a91a9832b899384c : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":7},{"id":"cca0904a8521d646d3211f568ce2867a8fe0b123","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:02:59.000000000","message":"Change has been successfully merged\n\n6 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":7},{"id":"876d76401d1c62a5599e83b1a963bd1ddbd1968f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:04:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b5791ff5264c5bbcbc9c4c2578c520\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3772f6b2ca4e6a9a73143e977074af : SUCCESS in 47s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"5c88c503490f235397d01a8c4dcac7004c2f0903","revisions":{"0faa1916a00e14ab94d013a59f9093419c899c1f":{"kind":"REWORK","_number":1,"created":"2023-11-04 22:36:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/1"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:36:04.000000000","tz":-420},"subject":"Improve object fields coverage","message":"Improve object fields coverage\n\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0faa1916a00e14ab94d013a59f9093419c899c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0faa1916a00e14ab94d013a59f9093419c899c1f"}]},"branch":"refs/heads/master"},"9ff5daab1219711784b5f9ddf42a09047e1cc66a":{"kind":"REWORK","_number":2,"created":"2023-11-04 22:38:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/2"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:38:22.000000000","tz":-420},"subject":"Improve object fields coverage","message":"Improve object fields coverage\n\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ff5daab1219711784b5f9ddf42a09047e1cc66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ff5daab1219711784b5f9ddf42a09047e1cc66a"}]},"branch":"refs/heads/master"},"0880ad14b1fb85f1c24a118e0d18aa841ed447fd":{"kind":"REWORK","_number":3,"created":"2023-11-05 00:29:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/3"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:28:09.000000000","tz":-420},"subject":"Improve object fields coverage and fixed error handling","message":"Improve object fields coverage and fixed error handling\n\n- Fixed bad error handling.\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0880ad14b1fb85f1c24a118e0d18aa841ed447fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0880ad14b1fb85f1c24a118e0d18aa841ed447fd"}]},"branch":"refs/heads/master"},"88aeb7edf65add916c9b6132e0e1952a8018931c":{"kind":"REWORK","_number":4,"created":"2023-11-05 00:45:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/4"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:45:42.000000000","tz":-420},"subject":"Improve object fields coverage and fixed error handling","message":"Improve object fields coverage and fixed error handling\n\n- Fixed bad error handling.\n- Removed unreachable code.\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88aeb7edf65add916c9b6132e0e1952a8018931c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/88aeb7edf65add916c9b6132e0e1952a8018931c"}]},"branch":"refs/heads/master"},"3ff87d7a8dd6dfb1b1a0b3540779b3375108c2b9":{"kind":"REWORK","_number":5,"created":"2023-11-05 00:49:31.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/5"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:49:19.000000000","tz":-420},"subject":"Improve object fields coverage and fixed error handling","message":"Improve object fields coverage and fixed error handling\n\n- Fixed bad error handling.\n- Removed unreachable code.\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ff87d7a8dd6dfb1b1a0b3540779b3375108c2b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3ff87d7a8dd6dfb1b1a0b3540779b3375108c2b9"}]},"branch":"refs/heads/master"},"47bb6cf5ef614c731f2b01ee95081436b4d8c49b":{"kind":"REWORK","_number":6,"created":"2023-11-05 00:50:27.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/6"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:50:20.000000000","tz":-420},"subject":"Improve object fields coverage and fixed error handling","message":"Improve object fields coverage and fixed error handling\n\n- Fixed bad error handling.\n- Removed unreachable code.\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47bb6cf5ef614c731f2b01ee95081436b4d8c49b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47bb6cf5ef614c731f2b01ee95081436b4d8c49b"}]},"branch":"refs/heads/master"},"5c88c503490f235397d01a8c4dcac7004c2f0903":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-08 00:57:43.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/900103/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/900103/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/900103/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/900103/7"}}},"commit":{"parents":[{"commit":"30d8c7081405f67aa8586194a39da2defa566014","subject":"Disable tox sqlalchemy tests until fixed in oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30d8c7081405f67aa8586194a39da2defa566014"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-04 22:33:37.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-08 00:57:43.000000000","tz":0},"subject":"Improve object fields coverage and fixed error handling","message":"Improve object fields coverage and fixed error handling\n\n- Fixed bad error handling.\n- Removed unreachable code.\n- Minor cleanup for consistency.\n\nChange-Id: Ie632f3c1358a6d25a75dfc023fa053b89b32ebf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c88c503490f235397d01a8c4dcac7004c2f0903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c88c503490f235397d01a8c4dcac7004c2f0903"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
