)]}'
{"id":"openstack%2Fdesignate~900104","triplet_id":"openstack%2Fdesignate~master~I5f4948d5cc03e2e6ee2ce2a1bdb47f4c408c53ac","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-08 11:37:06.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I5f4948d5cc03e2e6ee2ce2a1bdb47f4c408c53ac","subject":"Use super().setUp() for APIv2AdapterTest","status":"MERGED","created":"2023-11-05 00:51:07.000000000","updated":"2023-11-08 14:05:54.000000000","submitted":"2023-11-08 14:04:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900104","meta_rev_id":"0d14b0f1ace5a2282408b39168371b6f7dadd35e","_number":900104,"virtual_id_number":900104,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 14:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2023-11-08 11:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-08 11:37:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-08 11:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-11-08 11:37:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-05 01:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 01:05:28.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-08 01:00:17.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-08 08:59:22.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"f0c557dbacfbcf4b02ca72fce826df810515947a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-05 00:51:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cedbe5375524d69e652fb2a2d1c35c09d3f077fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-05 01:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73946b3db174ffd80aa54f399b70c52\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7743cf26995b43dbaa0764f8ff6a3f09 : SUCCESS in 30m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b5a854ab939f491185f7a036faab1d86 : SUCCESS in 38m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/100d8bc0e93b42fd8c5894a0685b77c6 : SUCCESS in 36m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e40e51ff14bd48099f2d4c4ea99f264c : SUCCESS in 32m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9e5b3e9aac604a65ad936c04e43913ad : SUCCESS in 47m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee2a63c6dcc243db83c8666f25895cd5 : SUCCESS in 47m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/102a935eddfc41979b8354deed982156 : SUCCESS in 1h 06m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0d0aa562fdcc48a48a954c069e8be7a7 : SUCCESS in 29m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aa4b333d50f24ea8b0c7d7206bdea0fb : SUCCESS in 52m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/341a9e6de1434afa981e7d553c1cf484 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a39ece75fc5408fa555c1cc7432f299 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cf3e5e5216d40179c74aa493d275acf : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec610caa7a1d4e7badc7e32a1132fc5d : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73ace7548f5446f911a343f4a11e9f2 : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/914b884c7fdd4a739028f18f85ae47bf : SUCCESS in 33m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eea8fd14cef542fe9a21df9ca38d49a3 : SUCCESS in 5m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/520b9a613ee5406aa9782ad6c0627716 : SUCCESS in 32m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/81e4af5395114d40a7402993a2853a72 : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60fb9ef29637f42a3f3d1b8eb535c270db0527d8","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-06 01:05:28.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e7e5bedbef5410cbc434c1bc0a43ad010bf1e868","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-08 01:00:17.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e664b9052aa977e639e71625f8eb7554bb0e083","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-11-08 08:59:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"88d393789c520e020996de78054d22e62950f746","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 09:06:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f5ed8f7a7e13049e455d12cac9a11c20a78c2bd","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 11:35:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90b8de4bda6c98094394f8be7133a911c03f162c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 11:36:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3c23b21a60c3781fb82b18816f73c608d7e223f5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 11:36:08.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f96c9d86f7455f8d75de848965fb948c719505cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 11:37:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f21dcc5ad41fec0c3364f534156f0f0af34d3d43","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 11:37:06.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dd8ae4dd9e0f156f98a45c2dd2dc1be44524d720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 12:51:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d57de8ad01b40cc9d7c82d4289a90ac\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee5529c9cd394b42b1b2bba1fb9b614f : SUCCESS in 44m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/77f1f68f686f4e3ab5b60a01fda6a0f4 : SUCCESS in 1h 00m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e9d839ca466d4a3e8cf150f74afdbc42 : SUCCESS in 1h 00m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4b9c43718c4e45fda5b0e71af49b3ab0 : SUCCESS in 44m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cf80dd1d9b3b4706bef30f534d4ead12 : SUCCESS in 53m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a2625c931294429f8fb53e9d5a2b5ae6 : SUCCESS in 1h 08m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a2ec426cbb06436bba01093b303a2d36 : SUCCESS in 1h 00m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d37e5e3b2cc4fb0969d6d9692339fb5 : SUCCESS in 34m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b0ebcd01bfdc46ec9032026d408999f5 : SUCCESS in 49m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3859cff2f56345afb0a6d5938c89308d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f54860718bd44b2934ba0fd156e3175 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55f00225672a4dd4b186149036e79363 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ea3f38f17c547f8a9d29299f87a8d26 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b9512865394af0817b544915cb2294 : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/22336069214f40ad9e3fc51677159f2e : SUCCESS in 31m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/32ee306272bc44eea5e12364538b547d : FAILURE in 5m 25s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f147af625584097944559c6b3e66151 : SUCCESS in 42m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8a0e5b81d891499eb47744e04dac12a7 : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3d2bd3a4449a049d8870ca7ce51b7df7406c7a80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 12:52:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e535cd4832980a7950e0b936e25ecf7594dd4328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:04:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/081cadbea1cc4103a335a6e6297e5fe9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/75e127e500fe40b7a769eed6fb61cc79 : SUCCESS in 44m 45s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2e02c0705f55482fb3151aa4ca169ed8 : SUCCESS in 36m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/65c6156fa2d44843b5828cdb1a80cd7f : SUCCESS in 50m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/22b706512c634a15853a81575b60eb6f : SUCCESS in 1h 07m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/493dd355659d43a2876e77707163c3d8 : SUCCESS in 44m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cb7fd9239748461fb0f2488fd49f1723 : SUCCESS in 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2266ffd117433080b1be034daddc58 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f6481a0a9564614a574adc1adfc35fa : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a406ecbecb8409091e0256b66e64b50 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b849b4b6fde4af6b7137e77027c26f5 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/713dc866b55249229ed0013d82300df7 : SUCCESS in 45m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0c0ae63284e94044b36fc5afbe103efc : FAILURE in 5m 45s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a9db7426a2054e1da6ca4ff45b139d5d : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":3},{"id":"ad1943350b0c07da2be1db5f57d5decc185544a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:04:53.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"0d14b0f1ace5a2282408b39168371b6f7dadd35e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:05:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175829a218784c9eb0f7583fb1b1a2d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9399b89126ca4614927ccdb723f787f6 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b1dfe627f8c9c7e185e5d87d2e8d90b51e36cc88","revisions":{"4285e6d7d8a5fc7c0a38ad7029bfe9446ea414eb":{"kind":"REWORK","_number":1,"created":"2023-11-05 00:51:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/900104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/900104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/900104/1"}}},"commit":{"parents":[{"commit":"47bb6cf5ef614c731f2b01ee95081436b4d8c49b","subject":"Improve object fields coverage and fixed error handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/47bb6cf5ef614c731f2b01ee95081436b4d8c49b"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:50:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:50:57.000000000","tz":-420},"subject":"Use super().setUp() for APIv2AdapterTest","message":"Use super().setUp() for APIv2AdapterTest\n\nChange-Id: I5f4948d5cc03e2e6ee2ce2a1bdb47f4c408c53ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4285e6d7d8a5fc7c0a38ad7029bfe9446ea414eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4285e6d7d8a5fc7c0a38ad7029bfe9446ea414eb"}]},"branch":"refs/heads/master"},"206c81efb681a0fab97fdefe3bc9bbb269b8d3cd":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-08 11:36:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/900104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/900104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/900104/2"}}},"commit":{"parents":[{"commit":"5c88c503490f235397d01a8c4dcac7004c2f0903","subject":"Improve object fields coverage and fixed error handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c88c503490f235397d01a8c4dcac7004c2f0903"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:50:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-08 11:36:05.000000000","tz":0},"subject":"Use super().setUp() for APIv2AdapterTest","message":"Use super().setUp() for APIv2AdapterTest\n\nChange-Id: I5f4948d5cc03e2e6ee2ce2a1bdb47f4c408c53ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/206c81efb681a0fab97fdefe3bc9bbb269b8d3cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/206c81efb681a0fab97fdefe3bc9bbb269b8d3cd"}]},"branch":"refs/heads/master","description":"Rebase"},"b1dfe627f8c9c7e185e5d87d2e8d90b51e36cc88":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-08 11:37:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/04/900104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/04/900104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/04/900104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/04/900104/3"}}},"commit":{"parents":[{"commit":"9328297cbe52c2b1edd65d8ccc69a86e266d8484","subject":"Fix coverage issues with greenlet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9328297cbe52c2b1edd65d8ccc69a86e266d8484"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-05 00:50:57.000000000","tz":-420},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-08 11:37:01.000000000","tz":0},"subject":"Use super().setUp() for APIv2AdapterTest","message":"Use super().setUp() for APIv2AdapterTest\n\nChange-Id: I5f4948d5cc03e2e6ee2ce2a1bdb47f4c408c53ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1dfe627f8c9c7e185e5d87d2e8d90b51e36cc88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1dfe627f8c9c7e185e5d87d2e8d90b51e36cc88"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
