)]}'
{"id":"openstack%2Fdesignate~900106","triplet_id":"openstack%2Fdesignate~master~Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-15 23:33:21.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a","subject":"Fix bug with new zone default values being ignored","status":"MERGED","created":"2023-11-07 14:09:57.000000000","updated":"2023-11-16 00:33:07.000000000","submitted":"2023-11-16 00:31:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900106","meta_rev_id":"1b3aca238523d829bb752e5944f211ff1916d756","_number":900106,"virtual_id_number":900106,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 00:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-15 15:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-11-10 20:12:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":1,"date":"2023-11-13 06:02:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-15 23:33:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-13 11:49:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 14:20:50.000000000","updated_by":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"reviewer":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"state":"REVIEWER"},{"updated":"2023-11-07 15:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 06:02:17.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-15 15:47:21.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6b5cc99b0a556ecf30463c9ffdf8d654fb1e551e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 14:09:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"504407c249702ef3a540b3239e8c7eb907af5cfc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 14:15:52.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"94f1ef4b5f0b90a09ff8ca0a276275492cdd8b53","author":{"_account_id":20870,"name":"Alex Kavanagh","email":"alex@ajkavanagh.co.uk","username":"ajkavanagh"},"date":"2023-11-07 14:20:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2a6a2946805403f11dcd8f803a523d9c1c5b933","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 15:18:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd82b2985ed34b29ac1c057bb5cd89ee\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/31405e2431324d8d9a4c94a820f31136 : SUCCESS in 40m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c7fc9926308f4c69a0dcd68081691c88 : SUCCESS in 51m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/31aa37fd8d1f441186a7453c468986aa : SUCCESS in 41m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c2d45dfb3c5e405da1938719f1be0df2 : FAILURE in 38m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/90d1638bfeed454998118581aec09688 : SUCCESS in 37m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6a8e9a22a28c4940bbd6d8e02a3d09b1 : SUCCESS in 1h 02m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a480038ac894ba28c563cc5ca2cc435 : SUCCESS in 55m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f8ab12ea8f5472a8b00c296ecef846e : SUCCESS in 42m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aeb41bb2448d4428b8d2bef1754d0b18 : SUCCESS in 41m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54292d22e9da4e36ae97b123af2e343e : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1566a8b9ea4afa83e294f40805d146 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/985bb4f58b8b4745b484ad8a209430b4 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1f27da91ec24618bc7d3fee150daa38 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96dae4cb63e497caf0d02534cc67c32 : SUCCESS in 7m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/340a61761ae84bce85c4dee7a2bdbd06 : SUCCESS in 43m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9ca371eb48fe4dbc9d3fd533e2ebb818 : FAILURE in 5m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/eb8cd570afa8491a819e88be7904e305 : SUCCESS in 54m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/872ec4020b9a4ab587c93a1c1b4f2d0b : SUCCESS in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"534d5cac445d2148e9d594beb764e77952e48c33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-07 23:32:59.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1559079441c595344b12f8a5e3b7797a193ef4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 01:01:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d5227d309e48d9a7032b295c34bf26\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3c6f3dcd3dcc425a8531f58fb97a11c9 : SUCCESS in 33m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5ae99bf2cec04686ae668043b2bb8927 : SUCCESS in 50m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee442c2ef81a4769a4f8ff5a4401db42 : SUCCESS in 58m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2197966e021e48fdb935d961b0e03878 : SUCCESS in 53m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/375c5275bbab48c0accbdea1c368d273 : SUCCESS in 49m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/883d972a56324e4b8604f10d88296b4b : SUCCESS in 1h 16m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3b9d5b45fe404c61beffe7070af6a37f : SUCCESS in 1h 17m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4c60914d9d834f38bc0bc1882ff7543c : SUCCESS in 52m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/03ca836204d34e24a325f3866ae2d8b3 : SUCCESS in 50m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91f5385fd2bf4ac7bfc7bb1ceb6281ec : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77994bfe53041028c95ebca30ff122d : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2973676d864f48139638150e150b5b71 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08d4bf6da7824853ae5e894bd679c498 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0099206339594cc2aea4814d5fa8e040 : SUCCESS in 7m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/18cdcc993b104ac7a0d94f9b349f9ede : SUCCESS in 34m 24s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d82e9398b3cb4db099e1c3175bebb5c0 : FAILURE in 5m 57s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/66303a7ee23743ff93e36475fbe3eb4f : SUCCESS in 53m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ca3b2d8c7f36440b90812b5343e6a426 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6c0c3ca7870c6e484c4f76bd8ec5438ef58f077c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-10 20:12:45.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"663cb192591455be2602ec118abc838782862c18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 21:27:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2edbcdcc1fbe4c01b2b03afaf7b5c713\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fd99d54c9f064a279c31123d365fdd55 : SUCCESS in 43m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/658cb1bedd3047fba43cdab3e95447c2 : SUCCESS in 45m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7e97ab8143f4f8598bd09f9dd521e81 : SUCCESS in 43m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a15b01ec04d34876b9fd252e9669ad46 : SUCCESS in 34m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/58ac81772d1742a7897024788dee2c1b : SUCCESS in 51m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/518a97ab6a044aa4b9b8078cf8369658 : SUCCESS in 57m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f580d361ecf044b7bc991f1b4e955e43 : SUCCESS in 1h 08m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58152f1bce394e6a937c3a9772a6f3ae : SUCCESS in 48m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7d969aaf02984ba8aeec7fb80ce2e727 : SUCCESS in 47m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b57bc6ba4090479c87b32a319b0a74a1 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1c699bdda54e33bbacd8d897ec0885 : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e78a9a8b5f9436a8de55961c30d0c61 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16ea7f324f6641619a1bdd17039d694f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7660667d25042b28cc9ca3dc29a86a1 : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71d24cd8091c461eaf50d8ce72be3c88 : SUCCESS in 33m 00s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b000434d8e694a52b3e2a0f8ab306ac6 : SUCCESS in 5m 45s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2af88fc35f8142fd8eb245e27d05006d : SUCCESS in 36m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a671002088224bd5a976229eb8255142 : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3dd1314aee1697bee24e9378ff207504d14b24f7","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-13 06:02:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b3a9ae0a1bf01d8ea9a0fe599824fa83f9af91f2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-13 11:49:37.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"160eae74fc3bc43b1ede8499ae77199af1676136","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-15 15:47:21.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d8328c4235b121ff9d572b34b2b4edc6ba9b9c38","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-15 23:33:21.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"69b9ea70bdc9636e21a9bf7d4cc5bd37a2fe284c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:34:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8facd898bea7d6dac12893745d4ae729d36da40b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 00:31:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938a082ef7594b3d97c56abd01056d84\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5e8e2a6eee444753bffe8a74bd85eeaf : SUCCESS in 43m 55s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1586b3c5d51345b19cd2f4066ec0ffde : SUCCESS in 32m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c1e6312a604d48858ae4b5f6aeed9a03 : SUCCESS in 31m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a173deee4705437289db86d500071a33 : SUCCESS in 49m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f66ca9c3715b48e0a262a8319443e72c : SUCCESS in 34m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/06b78abc1be14129b97189a4e852347c : SUCCESS in 32m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02145e4a7f3e48f9bad1e7a569765dde : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8d844a6746d4aa9aeeebec19780e49c : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/830e7a74b3bc4d6b9bcfda7422c7c3a2 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c6e7348776458b9d908b4c51ad1e8d : SUCCESS in 7m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c47f853ce9a4507a4e2bda3fae193ad : SUCCESS in 39m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c203f2ca457b454ab3b920d5e7f90096 : SUCCESS in 6m 47s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bfb0a62fcd794622a1e784be915386eb : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":4},{"id":"90df6ce2369692c819171925d821af627236a741","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 00:31:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"1b3aca238523d829bb752e5944f211ff1916d756","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 00:33:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5df26319bca44e0fbd15151ae32a3c3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6dedf304bc4d13bf4e93a6ca67c1e3 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"aae83d09b88c90a5f0872daa206dbf5921aa1827","revisions":{"13ceb9af58c099d6279a0b1921e38447c3de5ff3":{"kind":"REWORK","_number":1,"created":"2023-11-07 14:09:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/900106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/900106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/900106/1"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:08:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:08:37.000000000","tz":-480},"subject":"Fix bug with CONF.default_ttl value being ignored","message":"Fix bug with CONF.default_ttl value being ignored\n\nCloses-bug: #2042944\nChange-Id: Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13ceb9af58c099d6279a0b1921e38447c3de5ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/13ceb9af58c099d6279a0b1921e38447c3de5ff3"}]},"branch":"refs/heads/master"},"132cab251428f0e5fce2afda39d9b1a5af7178e0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-07 14:15:52.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/900106/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/900106/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/900106/2"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:08:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:15:52.000000000","tz":0},"subject":"Fix bug with new zone default values being ignored","message":"Fix bug with new zone default values being ignored\n\nThe following defaults weren\u0027t being applied properly\nto the zone.\n- default_ttl\n- default_soa_refresh_min\n- default_soa_retry\n- default_soa_expire\n- default_soa_minimum\n\nCloses-bug: #2042944\nChange-Id: Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/132cab251428f0e5fce2afda39d9b1a5af7178e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/132cab251428f0e5fce2afda39d9b1a5af7178e0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"870f4358253329ce387eea0d6b74bf564c6145a3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-07 23:32:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/06/900106/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/900106/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/900106/3"}}},"commit":{"parents":[{"commit":"30d8c7081405f67aa8586194a39da2defa566014","subject":"Disable tox sqlalchemy tests until fixed in oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30d8c7081405f67aa8586194a39da2defa566014"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:08:37.000000000","tz":-480},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-07 23:32:59.000000000","tz":0},"subject":"Fix bug with new zone default values being ignored","message":"Fix bug with new zone default values being ignored\n\nThe following defaults weren\u0027t being applied properly\nto the zone.\n- default_ttl\n- default_soa_refresh_min\n- default_soa_retry\n- default_soa_expire\n- default_soa_minimum\n\nCloses-bug: #2042944\nChange-Id: Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/870f4358253329ce387eea0d6b74bf564c6145a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/870f4358253329ce387eea0d6b74bf564c6145a3"}]},"branch":"refs/heads/master","description":"Rebase"},"aae83d09b88c90a5f0872daa206dbf5921aa1827":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-10 20:12:45.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/06/900106/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/900106/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/900106/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/900106/4"}}},"commit":{"parents":[{"commit":"4e185d4e266299ce22376024977cd9471dcaa0ed","subject":"Merge \"Make sqlalchemy-2x tests voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e185d4e266299ce22376024977cd9471dcaa0ed"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:08:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-10 20:12:45.000000000","tz":0},"subject":"Fix bug with new zone default values being ignored","message":"Fix bug with new zone default values being ignored\n\nThe following defaults weren\u0027t being applied properly\nto the zone.\n- default_ttl\n- default_soa_refresh_min\n- default_soa_retry\n- default_soa_expire\n- default_soa_minimum\n\nCloses-bug: #2042944\nChange-Id: Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
