)]}'
{"id":"openstack%2Fdesignate~900108","triplet_id":"openstack%2Fdesignate~master~If6a3ab5f762453c50f23b668328cf20d930d6b01","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-08 09:07:05.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"If6a3ab5f762453c50f23b668328cf20d930d6b01","subject":"Fix coverage issues with greenlet","status":"MERGED","created":"2023-11-07 15:39:21.000000000","updated":"2023-11-08 11:34:58.000000000","submitted":"2023-11-08 11:33:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900108","meta_rev_id":"e793d82bac952018b7f23f9a24bdc0def75bbb11","_number":900108,"virtual_id_number":900108,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 11:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-08 00:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-08 09:07:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 17:18:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 00:59:02.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ea59b3ecbce0775414c520e95ceefb0978272974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 15:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"820c260c4ada2b7e28736ec209d0d90ef5d212b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 16:00:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0603b6622f06311ab592603edcdf636fa69a70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:18:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a35e9961f730436383562bf002a324b7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f59eea10671d4459b35ddca2b321a4fb : SUCCESS in 53m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/97d75ee8685047bcb5ceb9cc1f93050b : SUCCESS in 45m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a15000ba2c342cca6f01e0e90757e6a : SUCCESS in 1h 04m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f04e9e3772a44aa987c06a9fde7911bd : SUCCESS in 34m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/949f22ac441049d182758b85d73581aa : SUCCESS in 49m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cd97b3adad4948aabc7452aea8227c70 : SUCCESS in 50m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ea96e06870a44e36978e9df9ef9de3aa : SUCCESS in 1h 12m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/65ed41a765d24d518d6a2237fc61131d : SUCCESS in 54m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/586c7fe5f0cf4b459501baa5a23213d8 : SUCCESS in 36m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72e946312fb49618ba14c2cd2cf447e : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c9487505b242d88a6e6a94d5903109 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8504ce4f9193428e9bc4d5943620ed71 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b08e5ac29ad46a29a48a77a74ef55cf : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26735ccb25a146f7962557b4748ac325 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e67e22fb479c4896b448f365f45642f9 : SUCCESS in 41m 10s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8ad9d68ee6c747ef939fa0a80b981781 : FAILURE in 5m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b25783c0379d4e6886e858e07fa7c9b3 : SUCCESS in 46m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a49e5fce75e246c9a91828b0a48a419f : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04e2d562d8137ac5989ada451f0bdd0e1b406a87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 00:56:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b5995d4c0bcbf82116633d11d021053f269cb19","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-08 00:59:02.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"357813997174cfb303e64eeeade0209aff7b331e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 02:04:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d6e378060c74984a7d070fefae8ec75\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/443e551438ff42a2b9867f0f5e4a04eb : SUCCESS in 47m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4ccc2bdb5e4d45cbac7561b47d4d3089 : SUCCESS in 1h 02m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a71f35bd0c94041b023de58083a3d41 : SUCCESS in 56m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/60f443b059624c78ab6543a7b464b8d5 : SUCCESS in 31m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/70e4bcea50ef49779dcbf6f8c876588f : SUCCESS in 47m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/532b6cbe64eb4b298f3188f9429bc081 : SUCCESS in 51m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e42115c5269d4e4d91f8523d8a9f3d5a : SUCCESS in 51m 40s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b8eac8a2cd0245459c58ba225e9fd2f9 : SUCCESS in 40m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/017d608955474cce89408764b998e9b0 : FAILURE in 52m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/439640097eda4f07ab95634fd7d2b0f1 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/354056306ab0409f86771df8bd349c9e : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2d7282d6b1c47c5b47b8787435e5c23 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25d2e3bf2d5b4dcca172c8f0a60fa64e : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0b7df8c67e4b33bd469369a7ad623d : SUCCESS in 9m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9aa9e87eb1c64262a9c6ff7671af5ea8 : SUCCESS in 41m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b9284eb282e040de93f471262bfcb052 : FAILURE in 5m 25s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/21e7ae6da23947078d356478f6fae1e8 : SUCCESS in 51m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/87a47c7b31b240818d27238bc33a40c0 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5b605da48b21766cb4a95f1e779f2312bfb11d20","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 09:07:05.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70a409cc247954c8820fff1b4eea449c68b665d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 10:29:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1409deefb8e240e5b3575b0a82377d57\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0fb34987f45842a7b9ce23dc6bd84eb9 : SUCCESS in 36m 28s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9c2a1987eb3742368be61dc01d66c40c : SUCCESS in 45m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd4fccdfa25846e48434e3c99eadeb30 : SUCCESS in 38m 39s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1b5fd0c14d3a44c384f32d93754325cf : SUCCESS in 36m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9b8d02c33897497f8155c7670ba6008b : SUCCESS in 49m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a6c5e98442ed4e28aef7b8a1e34910cd : SUCCESS in 1h 08m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a363291eae7f4fe09b7942a62abd1b7f : SUCCESS in 1h 14m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/62f2348cd954439ab3d436d69b88e37f : SUCCESS in 48m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5f5e204a427d47728960c89036abc575 : SUCCESS in 49m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c68c8e1655f14895808a3d28b0eeeec5 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5b826ecdae42f48e6c0b0f51dcd2ad : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16abd968a4744d658806aef38afff93b : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa7ece55c25b46068a2eaee5ba1c5b43 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed10ed9554349d492c34c257f09e4c8 : SUCCESS in 8m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7a4e88711da84636be41af6d59cc8139 : SUCCESS in 34m 02s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2142b75f64b4424f9cbc3311da6dac64 : FAILURE in 5m 50s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a661b00efe894f6096acb572b4138ac5 : SUCCESS in 51m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/87f2353a706849009fbd2102fe00b11a : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"133b374673942c3210754312345c15a1a450d3c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 10:30:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aaa8dbb9d4ae8313984c963350e9a4373e129179","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:33:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc96e0525c04b60abf6855cc79f12f8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0ce73be6b654c9facbdd479988c85ed : SUCCESS in 38m 37s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/88c435eca50b497e95aa627bd404487d : SUCCESS in 35m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9f93b679ca4846aea0ca3e5997c33475 : SUCCESS in 50m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/24f2079c0d214018aa90bf7bd7b01945 : SUCCESS in 1h 00m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b941a6d4b689480690cf8e6b0af01f9e : SUCCESS in 37m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3e14ed202bdc41d68b4c27feae55c35e : SUCCESS in 37m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bcd60b323df4297bc1b962e1a69a4c7 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a9f777793aa4a0586b8b55de0ff5d1d : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70164bfd9bbe4dbeb49406bf8e397f21 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c606b352fb4144acb771f6cf9ec92b3e : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/802a39c992784d1c813bc2d2588a0027 : SUCCESS in 41m 01s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d62ce44880534cae8459639022b56984 : FAILURE in 7m 20s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bee950ec8b5242eebf08cb4fcb2615bc : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":3},{"id":"d4b8ac52eb7475ec1381b2b4aa5a0b146787d9c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:33:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e793d82bac952018b7f23f9a24bdc0def75bbb11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:34:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cac2be05527645fdb99c9c3e36e6f85a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1fadcbd14c34c1f90deac262d0b3a08 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9328297cbe52c2b1edd65d8ccc69a86e266d8484","revisions":{"fe37e7b69b578e45b254d013ec075fade8c06660":{"kind":"REWORK","_number":1,"created":"2023-11-07 15:39:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/900108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/900108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/900108/1"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:39:06.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:39:06.000000000","tz":-480},"subject":"Fix coverage issues with greenlet","message":"Fix coverage issues with greenlet\n\nChange-Id: If6a3ab5f762453c50f23b668328cf20d930d6b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe37e7b69b578e45b254d013ec075fade8c06660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fe37e7b69b578e45b254d013ec075fade8c06660"}]},"branch":"refs/heads/master"},"ac57056f1056c79b484e2f969e2527f7440f487b":{"kind":"REWORK","_number":2,"created":"2023-11-07 16:00:30.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/900108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/900108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/900108/2"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:39:06.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 16:00:22.000000000","tz":-480},"subject":"Fix coverage issues with greenlet","message":"Fix coverage issues with greenlet\n\nChange-Id: If6a3ab5f762453c50f23b668328cf20d930d6b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac57056f1056c79b484e2f969e2527f7440f487b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac57056f1056c79b484e2f969e2527f7440f487b"}]},"branch":"refs/heads/master"},"9328297cbe52c2b1edd65d8ccc69a86e266d8484":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-08 00:56:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/08/900108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/08/900108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/08/900108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/08/900108/3"}}},"commit":{"parents":[{"commit":"30d8c7081405f67aa8586194a39da2defa566014","subject":"Disable tox sqlalchemy tests until fixed in oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30d8c7081405f67aa8586194a39da2defa566014"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:39:06.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-08 00:56:11.000000000","tz":0},"subject":"Fix coverage issues with greenlet","message":"Fix coverage issues with greenlet\n\nChange-Id: If6a3ab5f762453c50f23b668328cf20d930d6b01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9328297cbe52c2b1edd65d8ccc69a86e266d8484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9328297cbe52c2b1edd65d8ccc69a86e266d8484"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
