)]}'
{"id":"openstack%2Fdesignate~900109","triplet_id":"openstack%2Fdesignate~master~I6afafb225a970777dfff6447d42d1ac027b4795a","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-08 00:53:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6afafb225a970777dfff6447d42d1ac027b4795a","subject":"Disable tox sqlalchemy tests until fixed in oslo","status":"MERGED","created":"2023-11-07 15:55:50.000000000","updated":"2023-11-08 01:01:12.000000000","submitted":"2023-11-08 00:53:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900109","meta_rev_id":"218bdcc18255466996afaf7503e58ef8fc477abc","_number":900109,"virtual_id_number":900109,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-08 00:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-07 23:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-11-07 23:32:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 17:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-07 23:32:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6cfe8064508eed6d5c0409c19f37fe45b1950547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 15:55:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6950b42b603ae9179009720acfd6dff4674a4478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 16:21:07.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9d140b3d84f5beb962aa9c03244e28e41f4f20e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:33:52.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9d2e46b29a340ea82d09bec7aa3b4cc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a987ace742764dd4a54ad873f98b3115 : SUCCESS in 50m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5ab9371e88f546dca11acc92b39011e3 : SUCCESS in 50m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/25e0f0eef5704acbb271ac5a138346f9 : SUCCESS in 59m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f48b2daea0034cf28dd2af1316e21242 : FAILURE in 14m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f35a59ddfbe74bee811b61c25607a747 : SUCCESS in 50m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5f1dce567fac4b4db58d15b9f369b8f3 : SUCCESS in 59m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ae6d7a20c5a4669937a9e0bf0186c86 : SUCCESS in 1h 07m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8f4327070277406ea721b7c842b59169 : SUCCESS in 36m 31s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9be2c87762fa4708ab8a059070479f9c : SUCCESS in 50m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6f2ebdf757a47fa8ff28a635debe5f0 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb8686d83cd42979cb7c4a8e545f858 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34e880b4b1a943f187567babbe0ec5ce : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e3894ab45c346a3bb1e54728c1137f6 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97bd6d8ae998465eac047af6b22b5ba2 : SUCCESS in 8m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b217df363004112b2c98f0deb563f8d : SUCCESS in 33m 13s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/238e391bf851430db96feb4fb5c2b0bb : FAILURE in 6m 33s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3d5b6391ea5447bb98226c637fbca27b : SUCCESS in 48m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b3784f409abe4f6cb010d95214ca3f34 : SUCCESS in 4m 32s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":2},{"id":"f343e7ae29cff3de4c386302b0c34d206455df47","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-07 18:20:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd161a79a663000f1a8acbb8d7708c9db4419c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 19:20:55.000000000","message":"Patch Set 2: Verified+1\n\n(2 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d05d2218e1584c618e2b3b0ad6b654a4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e7e62896585448faa0c2ff1247eec401 : SUCCESS in 39m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/05235752166f48f7a89b3d9b395a97b7 : SUCCESS in 44m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/66ef472a6cee43e99b154a5bcf4983f5 : SUCCESS in 43m 01s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7d72c977f9d148e696aa3ab554ed4b55 : SUCCESS in 37m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c06af9f9561340edb0d88aa404b74f40 : SUCCESS in 32m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fed188f7e65049a7811ea3dce5b0e820 : SUCCESS in 56m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a862843e2c684dac8df44c098e6de1e3 : SUCCESS in 58m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f857695e140148dea72ea227dc4e74c6 : SUCCESS in 40m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1c8600f592754092902167f108782c98 : SUCCESS in 35m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0364cf9f29374d9799d217186ece06a1 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53b2b3fcd89a454facc6107f04a92002 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/769f22201cf5453db2515411e5a6da55 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f29648eefc8a46d1b55f721df1a4e74b : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fa1bd6557ff40dd961125ea8e75adf3 : SUCCESS in 7m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b25cc8974534cbdb8703a5a5ecd9089 : SUCCESS in 38m 21s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/82a3329bc47d479b980a4229382a8ad3 : FAILURE in 6m 09s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e299d8062927457ead75b028d0054bee : SUCCESS in 33m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e7608f42e14c489cb6b6f07f9da7d41d : SUCCESS in 4m 31s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":2},{"id":"54954f557a75a4982dad25f2a1cb2a406655492f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-07 23:32:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df88a9c1be67b84f01746fa80c965b1eba251df3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 23:32:55.000000000","message":"Patch Set 2: -Verified\n\n(2 comments)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":2},{"id":"fbf3f41c15393102f83192a624bf789299e9494e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 00:53:43.000000000","message":"Patch Set 2: Verified+2\n\n(2 comments)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44108527c1b942ceb177c42c6ee50b12\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a315f9163d92425c9bd874316500c38f : SUCCESS in 37m 24s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4cdd47afe1f549c2b9050561fe92b5e1 : SUCCESS in 34m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/343e349b6cdc465ca4d30f83564aaa0d : SUCCESS in 58m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/50b46e7e6e3448cca687ad7a06d853bc : SUCCESS in 1h 13m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3a3967df9bd4afd8252088f4f1b75ba : SUCCESS in 52m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a55405e02597485facde59f5814a2efb : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3de55ebfb34ce18ec4180805a781e2 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa543ec5a53d4a3d8d03712705728f9c : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e44cbaf9d04436eb71488f460c73ff6 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9692990a94b249cabe931449167489e4 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/957a1b17a95b45a7b11647f5f190460a : SUCCESS in 35m 09s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/daa08de932654065922e8543d7cd5bc2 : FAILURE in 4m 58s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/17bf228aa60e4c5cb8234c393c807b63 : SUCCESS in 36m 09s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":2},{"id":"07d5537064962be2c85c9c1a6a953edff043da02","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 00:53:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"218bdcc18255466996afaf7503e58ef8fc477abc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 01:01:12.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841767f71c0f44ef9a43d2928a06d156\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb9dc32e1d84a3482e03df433dfb2ef : SUCCESS in 53s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30d8c7081405f67aa8586194a39da2defa566014","revisions":{"337dd49c24f661ce126408e36f3f6701d2287869":{"kind":"REWORK","_number":1,"created":"2023-11-07 15:55:50.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/900109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/900109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/900109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/900109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/900109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/900109/1"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:55:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:55:42.000000000","tz":-480},"subject":"Disable tox sqlalchemy tests until fixed in oslo","message":"Disable tox sqlalchemy tests until fixed in oslo\n\nChange-Id: I6afafb225a970777dfff6447d42d1ac027b4795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/337dd49c24f661ce126408e36f3f6701d2287869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/337dd49c24f661ce126408e36f3f6701d2287869"}]},"branch":"refs/heads/master"},"30d8c7081405f67aa8586194a39da2defa566014":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-07 16:21:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/09/900109/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/09/900109/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/900109/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/900109/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/09/900109/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/09/900109/2"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 15:55:42.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 16:21:07.000000000","tz":0},"subject":"Disable tox sqlalchemy tests until fixed in oslo","message":"Disable tox sqlalchemy tests until fixed in oslo\n\nRelated-Bug: #2042886\nChange-Id: I6afafb225a970777dfff6447d42d1ac027b4795a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30d8c7081405f67aa8586194a39da2defa566014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/30d8c7081405f67aa8586194a39da2defa566014"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
