)]}'
{"id":"openstack%2Fdesignate~900117","triplet_id":"openstack%2Fdesignate~master~I29cf34b5120cca2e8b14c9c9f8ffa7c6e58da5b0","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-09 20:35:46.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-09 21:12:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I29cf34b5120cca2e8b14c9c9f8ffa7c6e58da5b0","subject":"Fix producer unit tests occasionally failing","status":"MERGED","created":"2023-11-09 00:33:38.000000000","updated":"2023-11-09 22:43:05.000000000","submitted":"2023-11-09 22:41:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900117","meta_rev_id":"47cc09730f443ee47f7522b490f4ab4551551bc7","_number":900117,"virtual_id_number":900117,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 22:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-09 20:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-09 21:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-09 15:15:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 02:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 06:36:19.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c249a8b1b4cf62f6711c6082a17c5926933c3c38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 00:33:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59b642ce3c277bcd1489f75ddd04a1efccf6be34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 01:02:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"68f0bb93e4ebfd9b0b17cbc816e0ea14340db90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 02:14:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ed5e51581af49849470a6639d75ef8f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8f7216bc6c9440dfb9c98b61b49927ea : SUCCESS in 40m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3314f5327dc746a0879ad800838d1476 : SUCCESS in 1h 04m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/eed63e0b42474dfc861347344425a4a1 : SUCCESS in 41m 21s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5b2ae0227d5742d4ae02f3d7ebcd8216 : SUCCESS in 37m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/10f10de19f4842e891a674df6d1447a2 : SUCCESS in 46m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/227d3f0ca63e49439f651fc1baf35cb6 : SUCCESS in 55m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abc582fb75854235a0e719740f749ab4 : SUCCESS in 1h 11m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c596bf54e09a4e9f8562ede1d5a9bb4e : SUCCESS in 34m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9668839175504a8a8d58bd35a0069c96 : SUCCESS in 52m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90557464932b4d488074842ba67ce1cb : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abf51418bb61455f9140598a4956a28c : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fc22618a35d4fffa43c8f2ec6f08027 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b4f1102ca35486f8ac6802d4d593608 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce4ddfdc4254ff2b3b2e14fc9e44e4e : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/50433a6b5f6742468c9f1ccc5fe63b61 : SUCCESS in 34m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d4af6d4aeaff49debf04b0bca94c4a60 : FAILURE in 4m 58s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/60d9fcd09f5941daaef6f74d755e138c : SUCCESS in 39m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9fd4ed4b1cf749f8838e63f6f0e77b27 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9da2c7cbe1b4fac877289d282aed84bd27bac99b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-09 06:36:19.000000000","message":"Patch Set 2: Review-Priority+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"abe7163f50806a295a195027da3de62d217e983c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 11:19:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6a98afa211ccd2681e1e5e48d66301b4f83a5312","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 11:20:12.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"6234547df3c222fb75d630e584a4d9b7cbeab64a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 11:22:21.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"7a903c946df8927420737fbb2a01e54cdf79c98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 12:38:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ecd8678908f434f9828e5d7c11a9f0a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd2e613049e84fa9ad94cb237e202e9c : SUCCESS in 46m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f58fbcebdefd4add8bc29b8316b18283 : SUCCESS in 59m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/55bb1c0954014685a5497655d6105435 : SUCCESS in 39m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/112a4cbbcb574aacb4cfcebad74eff6a : SUCCESS in 49m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b7da7a3e5ddf496f98db667847d4c32a : SUCCESS in 46m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/57bcdbd1559048fd96e3b630356c3124 : SUCCESS in 55m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eadeb7191a644c36b4017cd91fb885ff : SUCCESS in 1h 11m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ddfb00fec35a45c9a9bdb75c8c052de3 : SUCCESS in 50m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/93ab3659fbd547cdb29e1a416466b866 : SUCCESS in 51m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54717ceba90e4bf5b69be7921abf13b6 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82dd66378ce4eebab3537dbcf0612a4 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b774ac222aa4479abca6aec911b43e1c : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e41f8f0471bc4c099a9e58fb769d7ef1 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/612cadf4212a40a7a6e39ac312fc2e1b : SUCCESS in 8m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6010ab8867a4404482b12a98683829ca : SUCCESS in 32m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e3be1168982a472fb08c6db05d3f88e2 : FAILURE in 5m 23s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/084bd39341934f5b92d79f792e6ee70d : SUCCESS in 43m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3f6f52e4fdd94fb7ae228207c5c8ed8c : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"111f356e8225e0dc89a0a14507d5ff0043639da4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-09 14:54:46.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bd2c9e5d1a36d80ad799efa69a2edc9424f34ec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 15:15:06.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"20731961d9333260fe43992dbe03895ce471b335","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 15:15:17.000000000","message":"Patch Set 4: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"32e12cc9dd8d61dc256abd16104939ce33872deb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 16:38:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f268fefde9c42b78beb1151e24e2816\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/941e44648a5e44d3b049e4f12568ba54 : SUCCESS in 38m 26s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/712201949fd8491ca104bbdecb54e45b : SUCCESS in 46m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9de6922ea5724537877375f66583c76d : SUCCESS in 55m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5240ccf2dffe43e9b17e839e5531705f : SUCCESS in 48m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2835a1bd69814d57856eca3a6f4fd0cb : SUCCESS in 47m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e6da34f8916d4fd0bba3236af08a7ed1 : SUCCESS in 56m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/28484fdd2eaa45d887436c57d9b81d36 : SUCCESS in 1h 14m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3c31ecf6fcf1455099eecd5d78289a61 : SUCCESS in 57m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/27c1af3ef65b4649928064688c3ac286 : SUCCESS in 33m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/319f3d4a6a6b4b70acaa7fef8a6bbb86 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd6347767ef4d23861767e9e9226a04 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a13e430620ae43cdb541dc328265c67c : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ca6b07539ae42e4ba541f20d7583862 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f9c17fdb39645c383b6a09202b95c0a : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dd7b70f94ca04fd5b79182bb9d6b204a : SUCCESS in 33m 46s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ca54f0876c784b00901f9b7ff0734a4a : FAILURE in 6m 31s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5ceb5ca4620e41e6b063b3ace4611ae6 : SUCCESS in 41m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fdac69027b0e4cabad76b305fa1ab7ac : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9331923a7957e5afc26eb8b94497aa651868ff2d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-09 20:35:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1b92eda3aaa3d9768d65c829f4d601dc6654a2d9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 21:12:54.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5eb72ce4291cd342e5eb90a5ca7042e9ad8707d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 21:13:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9460a16c06b40c473a3fb28bb3644a8fd473a7c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:41:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cebee4412f47d59559bad6f37b1ac7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3d717e1c73b740288c326266a8c04218 : SUCCESS in 35m 24s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/22dce231284d4460b37a6f274abb090e : SUCCESS in 52m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f57ef27c3c46490c9eb55667b482b044 : SUCCESS in 32m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/606eb9a4dc01499b9aff8b9d9d0dc77d : SUCCESS in 52m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab44f1b41e424c639e24d0ad4b12d218 : SUCCESS in 47m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/14b6e1b945604dddac925fe9cba68b99 : SUCCESS in 47m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca49b2cfaab4cc58c5029af9d7b803e : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf11c3fad7814caca84c3fd5bec6caad : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42cde51ee10e414792f127be6ceb0004 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843fe44d143c455ca02dad24c35bda9e : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/51fd592b1ac748868acbb62c26ee9776 : SUCCESS in 35m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3af48008308349c193e4b7830225d939 : FAILURE in 5m 42s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c7fed5ab93ae43018c7be08eca8eea7a : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":4},{"id":"4b15bd405b25db3bdaded73e0b147bc7fbaf1102","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:41:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"47cc09730f443ee47f7522b490f4ab4551551bc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:43:05.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7cbc4e313dd40558fa300671eb0c843\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91c97bff29c4c0c8c5dde780dd9f7fd : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"923f29664eefc0dfd1fa3c1e5773b174d45e1309","revisions":{"f3c02b7a8963ccd483447054a8b91f227960919d":{"kind":"REWORK","_number":1,"created":"2023-11-09 00:33:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/900117/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/900117/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/900117/1"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 00:31:39.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 00:31:39.000000000","tz":-480},"subject":"Fix producer unit tests occasionally failing","message":"Fix producer unit tests occasionally failing\n\nWhen these tests were implemented they were setup\nto test shards. The problem is that the range used\nin the tests ends at 4094, but there are 4095 shards\nin total. This means that if a zone is created on shard\n4095 it is always ignored in the tests.\n\nChange-Id: I29cf34b5120cca2e8b14c9c9f8ffa7c6e58da5b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c02b7a8963ccd483447054a8b91f227960919d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f3c02b7a8963ccd483447054a8b91f227960919d"}]},"branch":"refs/heads/master"},"dc6f47b7023b4599b1b03c346dfae7d3a5b796c7":{"kind":"REWORK","_number":2,"created":"2023-11-09 01:02:11.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/900117/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/900117/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/900117/2"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 00:31:39.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 01:01:58.000000000","tz":-480},"subject":"Fix producer unit tests occasionally failing","message":"Fix producer unit tests occasionally failing\n\nWhen these tests were implemented they were setup\nto test shards. The problem is that the range used\nin the tests ends at 4094, but there are 4095 shards\nin total. This means that if a zone is created on shard\n4095 it is always ignored in the tests.\n\nChange-Id: I29cf34b5120cca2e8b14c9c9f8ffa7c6e58da5b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc6f47b7023b4599b1b03c346dfae7d3a5b796c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dc6f47b7023b4599b1b03c346dfae7d3a5b796c7"}]},"branch":"refs/heads/master"},"cdf5857101d5654a9fc94979c7372dfe3088b17f":{"kind":"REWORK","_number":3,"created":"2023-11-09 11:19:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/900117/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/900117/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/900117/3"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 00:31:39.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 11:19:46.000000000","tz":-480},"subject":"Fix producer unit tests occasionally failing","message":"Fix producer unit tests occasionally failing\n\nWhen these tests were implemented they were setup\nto test shards. The problem is that the range used\nin the tests ends at 4094, but there are 4095 shards\nin total. This means that if a zone is created on shard\n4095 it is always ignored in the tests.\n\nChange-Id: I29cf34b5120cca2e8b14c9c9f8ffa7c6e58da5b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdf5857101d5654a9fc94979c7372dfe3088b17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cdf5857101d5654a9fc94979c7372dfe3088b17f"}]},"branch":"refs/heads/master"},"923f29664eefc0dfd1fa3c1e5773b174d45e1309":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-11-09 15:15:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/17/900117/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/900117/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/900117/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/900117/4"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 00:31:39.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 15:15:06.000000000","tz":0},"subject":"Fix producer unit tests occasionally failing","message":"Fix producer unit tests occasionally failing\n\nWhen these tests were implemented they were setup\nto test shards. The problem is that the range used\nin the tests ends at 4094, but there are 4096 shards\nin total. This means that if a zone is created on shard\n4095 it is always ignored in the tests.\n\nChange-Id: I29cf34b5120cca2e8b14c9c9f8ffa7c6e58da5b0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/923f29664eefc0dfd1fa3c1e5773b174d45e1309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/923f29664eefc0dfd1fa3c1e5773b174d45e1309"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
