)]}'
{"id":"openstack%2Fdesignate~900120","triplet_id":"openstack%2Fdesignate~master~I9709cdf3e7cf1efef52c92f1e116544a4cbe9b70","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-10 15:32:56.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I9709cdf3e7cf1efef52c92f1e116544a4cbe9b70","subject":"Make sqlalchemy-2x tests voting again","status":"MERGED","created":"2023-11-09 16:24:02.000000000","updated":"2023-11-10 18:01:10.000000000","submitted":"2023-11-10 17:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900120","meta_rev_id":"5b56602a0f65984c94ebfe920ca130e755720d2a","_number":900120,"virtual_id_number":900120,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 17:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-09 17:27:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-11-09 17:27:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-10 11:30:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 17:27:00.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-09 17:41:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f72049b87d50c0f4a76996edbc98da85b5e5748","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 16:24:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"854422de9831f648c14dacabda0e74d4e492a42c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 16:24:56.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"947c71661b4739d815c2be538f16a2a26795a8ed","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 16:25:43.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"b2947578204be54010faca50dc020958a8b73c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-09 16:27:54.000000000","message":"Patch Set 3: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fb1d9d66fc183a89ce21863cc58429dcec0a4099","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-09 17:27:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"31cb04988be5b466ddb19ea079836a6aa1446dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:41:02.000000000","message":"Patch Set 3: Verified+1\n\n(2 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d1ff3c847b403fba897703b17b3ec2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ff59a549ee4457bb567d1290d1fe669 : SUCCESS in 41m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2295b23def4349389abe2ceb99de8543 : SUCCESS in 1h 07m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1262a5faed1434eaf7efb59012b0bb1 : SUCCESS in 1h 05m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/dcf3065eea754b069a3305f0b0a5ec02 : SUCCESS in 36m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/77c212bb9c73421281a4d2832463b192 : SUCCESS in 38m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0c85957d22114737a65acb6e17a4f333 : SUCCESS in 58m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a36d9c496e7a4fe39733b571c74e7c8f : SUCCESS in 53m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f359a8db3503411e8f071836b80a8b0e : SUCCESS in 46m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/34c3f11a65cf45f6b1f2632f05b2125d : SUCCESS in 46m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c467f14cf16148e7805ce8e0d34de14a : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11f7efdc0ea04b0fa3475198423d20ce : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd9571f64b984139a35aa0ee809a4f0f : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/19f2c3b4b63844a983024428eb37f96d : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5dc7d2d87b44fdb229c2f5adccccc1 : SUCCESS in 8m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b92cf80cc21543a8893d49496dac0d30 : SUCCESS in 35m 16s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/22f3b4776a30444a85f347b38581bd04 : SUCCESS in 7m 13s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1818383a0a4f49d5b76edc6d1e9e4de4 : SUCCESS in 37m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ef858ff670a2436a932526e334095992 : SUCCESS in 5m 05s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":3},{"id":"e2cf736caf95d2d5a701d8300f1a36574f728a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:41:14.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1ead0162d984a15b18d77e104ce3969\n\n\nWarning:\n  Change 900517 in project openstack/requirements does not share a change queue with 900120 in project openstack/designate","accounts_in_message":[],"_revision_number":3},{"id":"a17b6403514c1b0b420a50ccf6754cd46f858792","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-10 11:30:54.000000000","message":"Patch Set 3: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c203bfb13133701f99593bf590188e6c87a45387","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:55:55.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb92154b8cc14d928548499fa2bcc804\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c43d410bf2714a7aa4342c265d69260d : SUCCESS in 44m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7e86b9a98898472398664b74c498a894 : SUCCESS in 48m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b6387c2509f4ab19e3c9f0b68b0e858 : SUCCESS in 56m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3b409adb376d43b4a186c9e835c84b4e : SUCCESS in 52m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1aca4b20abb742af85f9279de9faecc0 : SUCCESS in 36m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/78a68b22831342beb438ee196afb6069 : SUCCESS in 57m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5c65fe1b056442c19ff7be70302d6f35 : SUCCESS in 1h 16m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/140edda058184c36b052c55802454e23 : SUCCESS in 45m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c427ead34afa4aafa4e67617e998d155 : FAILURE in 41m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bda1ae363d24a16a45c9ca8fb904645 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804831a369fc4cb9bbd7667bafaa2917 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab1489e942ba49a7a0ee06233aa0c457 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/251672b583c44b8eb58f67ea0aaefec9 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afde22c57db140d7ab6b1b3669b20486 : SUCCESS in 8m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/505bfb2a7f6a425087c070053108b04e : SUCCESS in 32m 26s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e92ab20a467d4950a628a9659ce5925b : SUCCESS in 7m 07s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b2173750da0b42dbb8769d3f61e8e37e : SUCCESS in 50m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c1fa23cf6c6b41af96b2cd22ab51586d : SUCCESS in 5m 07s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":3},{"id":"447f31bb4f4d52f5d3c7078c315e4fcc5ecdb3e4","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-10 13:03:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1d2409e172c51ed4528d7b3e3032f17466c85fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 14:08:50.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0c8247ad82f4f6ba7d4dc3355870ab0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/756e2d92c8eb4f4ab2a10479406c6ab4 : FAILURE in 40m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a99d3db113bd430bae322f64ba98ace2 : SUCCESS in 44m 45s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1bc4b3f2d394bb6a53e85e7faa8795b : SUCCESS in 45m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d5dd830fab6d40529a355a6d0d1d8003 : SUCCESS in 35m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/527c5aa7312b4a1b8dd6499751c9d064 : SUCCESS in 37m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/390703cb3dfe4cefb08283e2ecdb5e3e : SUCCESS in 1h 02m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a946b704e11842f782dde7b560d4d468 : SUCCESS in 53m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d469b8d77ebd42c5b809a02d0afcec73 : SUCCESS in 39m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c2632690b3d644c4a6379a9b47210c18 : SUCCESS in 38m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/030d6f076c544e728bb01ca25225787c : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959e005e0af244ce89e52b453b3cb1e3 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f3460186b2438cb839f70cbd7572f7 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e63a12a2bbde4f10a5144bb3b0776ba4 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68351dab3184dedb7290219e2dd089d : SUCCESS in 10m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f68624ecda6b4648ada5e5dc07867f3b : SUCCESS in 43m 25s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4b06f19b172b464fb9ede38e39eae75d : SUCCESS in 6m 47s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0840d7e6dee647f2aa11671fef9bc9ff : SUCCESS in 50m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1c54bd15e596486d942fdd8b6e60e75d : SUCCESS in 4m 48s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/designate on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: designate-bind9-sqlalchemy-2x\n      parent: designate-bind9\n      description: |\n        Run tempest tests with main branch of SQLAlchemy, alembic and oslo.db.\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n      vars:\n        devstack_localrc:\n          USE_SQLALCHEMY_LATEST: true\n\n  in \"openstack/designate/.zuul.yaml@master\", line 83, column 3","accounts_in_message":[],"_revision_number":3},{"id":"66c45f45d1165553440dd348e6e30549eeff873f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-10 15:32:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d5405744e1413f338437fe1eab0173047e389b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:57:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bf9eb6e8b354594bc82a88958d8063b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ed6bb79185ea4367b0bc161f0c553e3a : SUCCESS in 48m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ad07112d73c046208107d8c9aa9336b9 : SUCCESS in 51m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5c05fb8ce4aa4b818208c6ca565ef909 : SUCCESS in 56m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e77358c06ae044969bd197f0ef1f673d : SUCCESS in 49m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/563f7f8ffcf94a54bf3fb1bc75c9d7b5 : SUCCESS in 37m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/78627de151214685a240cc5e55bd92ca : SUCCESS in 1h 02m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d83e42593666414a878df40f942a4674 : SUCCESS in 59m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b83fcdf56dc54882be3e6005c0e31d2c : SUCCESS in 37m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b66f5c318f3489ab1e0b6f66cf31f91 : SUCCESS in 46m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3b6c39ce8b54e878b4675f84db153bf : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f775981d38d4714b660cc9fe7aad180 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bb7af23bc9144d184613ebade18d900 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4bf186abb474ad2a972e80fd301370f : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8352698bb8bf4c57be9b36a82bd615f8 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e67a17d5369a46fab73bd96231f5faa2 : SUCCESS in 47m 42s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/28e87a1fe5fc466dbd38023cfca90dc9 : SUCCESS in 7m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/712e882b7d88487583e225da3d28539b : SUCCESS in 46m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/191c28e452724d90a404a326944b83e7 : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"28d766b036b65aeef69b1df1258af0dd1a1c565c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 16:58:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9673ee042429277a7ca36a586bad4275e5c0251a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:56:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58ccb7f675640b8ac9d50e350357021\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/de4286ac82ed47bbaff7a75ee104f23b : SUCCESS in 46m 41s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5d1790618eb948fe951e2855083929e4 : SUCCESS in 48m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9da889ad037d4205872b7b15cd090dd8 : SUCCESS in 38m 36s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/79ff3147235548d6aa804c3862dbc93f : SUCCESS in 57m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8b5eb37bcf4c41bca37668343b013343 : SUCCESS in 47m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/02e24a79a22348e79ae1dfc67ea90d6e : SUCCESS in 35m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3aba91107124cfa97977a9ae45894f8 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86faad7476694849a7e1fe0b0bea5e44 : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed368d162fb44720b796f871254d2b29 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d6b7a1af77488aaabeb2233bb0e3a0 : SUCCESS in 7m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ebd8661767e4f12b36114bac64cd73a : SUCCESS in 45m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0405c188789e481195422bee4f3d8297 : SUCCESS in 6m 14s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d553a9ba08174ce28aa2e150a7aea57c : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":3},{"id":"97f7457eab57d343e9f2099c2de8302581aee434","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 17:56:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5b56602a0f65984c94ebfe920ca130e755720d2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 18:01:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40061abb5d244d19a00496a2d32d9288\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d363d5fe3af349b38f865f771690a83f : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"16798494f3929648c449803bb88fcc5866ed17f6","revisions":{"62b4578f3656e2814146b1922a25e73a861130fe":{"kind":"REWORK","_number":1,"created":"2023-11-09 16:24:02.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/20/900120/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/20/900120/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/20/900120/1"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 16:23:47.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 16:23:47.000000000","tz":-480},"subject":"Re-enable sqlalchemy-2x tests","message":"Re-enable sqlalchemy-2x tests\n\nChange-Id: I9709cdf3e7cf1efef52c92f1e116544a4cbe9b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62b4578f3656e2814146b1922a25e73a861130fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62b4578f3656e2814146b1922a25e73a861130fe"}]},"branch":"refs/heads/master"},"f08a76af36a28448cf89c821e2335e2bfef8755c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-09 16:24:56.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/20/900120/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/20/900120/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/20/900120/2"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 16:23:47.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 16:24:56.000000000","tz":0},"subject":"Make sqlalchemy-2x tests voting again","message":"Make sqlalchemy-2x tests voting again\n\nChange-Id: I9709cdf3e7cf1efef52c92f1e116544a4cbe9b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f08a76af36a28448cf89c821e2335e2bfef8755c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f08a76af36a28448cf89c821e2335e2bfef8755c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"16798494f3929648c449803bb88fcc5866ed17f6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-11-09 16:27:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/20/900120/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/20/900120/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/20/900120/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/20/900120/3"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 16:23:47.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-09 16:27:54.000000000","tz":0},"subject":"Make sqlalchemy-2x tests voting again","message":"Make sqlalchemy-2x tests voting again\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/900517\nChange-Id: I9709cdf3e7cf1efef52c92f1e116544a4cbe9b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16798494f3929648c449803bb88fcc5866ed17f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/16798494f3929648c449803bb88fcc5866ed17f6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
