)]}'
{"id":"openstack%2Fdesignate~900378","triplet_id":"openstack%2Fdesignate~master~I7efbce549a3ee673431018635c2f568f46df0449","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-11-10 13:24:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7efbce549a3ee673431018635c2f568f46df0449","subject":"Fix Zuul warnings with regexps","status":"MERGED","created":"2023-11-07 23:38:33.000000000","updated":"2023-11-10 13:28:08.000000000","submitted":"2023-11-10 13:24:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"900378","meta_rev_id":"05fc71c9502a83546ad027b37ccf1761cf344d13","_number":900378,"virtual_id_number":900378,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 13:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-10 12:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-10 00:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-10 12:09:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 00:56:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 20:37:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-10 00:03:59.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-11-10 12:09:54.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ca60096043df333742c930f94f696e576dcdadf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-07 23:38:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e397135ed16ff2c8e17a9d44ef9e20ef4205879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 00:56:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2afa67701040428384f4f855be105c49\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6168dc66869a405f98461ef66fc4dd4a : SUCCESS in 35m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e832287d67f446f99eca368b6bcc1fe2 : SUCCESS in 1h 09m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6d8b1495f89428da10de1018883017a : SUCCESS in 55m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/78c511da93b54322875849d0a5897fde : SUCCESS in 50m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3364dc98ce9a4284be28c1d267866377 : SUCCESS in 51m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a2f7fcac75af4dd2aaf03a99f0bdf6f0 : SUCCESS in 1h 06m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f71c65fc3a648ba8342160f36808ce1 : SUCCESS in 55m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b59cc0bb49b8449eb5b3b13748f99e42 : SUCCESS in 37m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f85f80a20a3442b19144bfdd625ec5e5 : SUCCESS in 38m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0ee2692315743738a7a4af0f4e94ff4 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f814b7e9a26b47eba0ec3d147e8e72cf : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97564ba64a3c42a9ae329ba21946f7c4 : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/695036668b7f433497672091629d4c77 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1191ddf55c944a159b2239d261b44375 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0925531bd9bb49e18224896776c43895 : SUCCESS in 33m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/38325503966f44b9b731c20c79272306 : FAILURE in 4m 31s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0e9047a137964d4387d78f7d1596e401 : SUCCESS in 38m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/78ec725db46849a8ab5bde41c1783f4f : SUCCESS in 5m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cec6747cd8f82a88c85fdb51cd10da83ece5bcb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-09 20:37:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88f66b9a9dfce2ecfa10ebfa8223dd8bf0ca69fa","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-09 20:40:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"060d3aa05e906a04e81b1d9e61884fbc180b7466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece3ab7024d74e70a01834700d05a13f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/714b5adb2d5e477f987082530ed33d15 : SUCCESS in 48m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7b024aa8d1a840a28a5ef3ebe1728962 : SUCCESS in 1h 03m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/06636dceaaed480daa21aa701329d42c : SUCCESS in 55m 28s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/794714f33cbe44b29d4944a826811ae0 : SUCCESS in 31m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/048031f10cf04c7184653e8a382e2600 : SUCCESS in 51m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b29d522804974cdaaad78e06ed619abb : SUCCESS in 1h 12m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cbc408bc891d4b9fa5a7ba6a08bb6d5b : SUCCESS in 1h 06m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e118cb5852da4a339b3904b11cdbb71c : SUCCESS in 50m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ccb843d9b0e8420a8e8d86278583c236 : SUCCESS in 31m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bacb539f60d7474cbcd6878666f40b39 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fde7a21d21c4ceaacf650f429376356 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3922f838a4947a39d9e440e024eaf32 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e21e0b38771242a4b48dc3475d3bc4f8 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5eaf69dc13d4cf59625b6584a5d7c11 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/400ee8eb30ff41d9b97246f3d368a3a4 : SUCCESS in 34m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/27a3d19f3ef941f19247ca5e4a8c9274 : FAILURE in 5m 45s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7584f3c92f4349fe85532cebd36d0582 : SUCCESS in 34m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b0cfc4255000496b9c19beea39243659 : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f7963da6586362617711d79bb678746cad158c3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-10 00:03:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eccb68d728c7cad00267063757679fe021ef7a11","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-10 12:09:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"608937eaf80849d1d308e72158c285b382a2e9a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 12:10:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58d2e1650afd9583523358f22974d6e1de32517e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 13:24:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c74b160c572a469880f3acb5eb8d39a7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7648a4e0e2224626b4a480862ba0c867 : SUCCESS in 36m 05s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b8a0ccf344dc4be68476fec31232f5f5 : SUCCESS in 37m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd1a4a64054d4be994e3987749713851 : SUCCESS in 45m 58s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/735dd9857d7d4e8e99bf578f9027bdf4 : SUCCESS in 1h 08m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c51319bbf9d14b3bb6068305be02df4f : SUCCESS in 45m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ff26a7f886f048e6a6f47c4463f48cb5 : SUCCESS in 45m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904b1aa6729444cf9bfe38705054f9db : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab19325cb0ae454db64a876f05a5f6c9 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbb720b50a69491ead28e2374f4ba98d : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59401483d45b42d4876dd534ffcfe9bf : SUCCESS in 9m 19s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7932d80348344156abd8cb2646b5d1e4 : SUCCESS in 40m 56s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/beb6f41d9d12473b838c305fdd64394c : SUCCESS in 7m 07s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be6ba7341f784acdabbbbcf44eb0d52e : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"44c1313b1fd2ae1ed5f698a51f1b5c2a6c31f5b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 13:24:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"05fc71c9502a83546ad027b37ccf1761cf344d13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 13:28:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c6d9dfce10431386433e23df8d1273\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1ddf0ae30f64d00b94dc0f3584db0a4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35a66e3d7ee02cefa7cf80202a39dd4b9f95c9a0","revisions":{"35a66e3d7ee02cefa7cf80202a39dd4b9f95c9a0":{"kind":"REWORK","_number":1,"created":"2023-11-07 23:38:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/78/900378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/78/900378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/900378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/900378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/78/900378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/78/900378/1"}}},"commit":{"parents":[{"commit":"2eeba9ea7d526e022a5341acd4a367529b10ca2e","subject":"Increase minimum test coverage to 93%","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2eeba9ea7d526e022a5341acd4a367529b10ca2e"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-07 23:37:35.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-07 23:38:31.000000000","tz":0},"subject":"Fix Zuul warnings with regexps","message":"Fix Zuul warnings with regexps\n\nzuul now reports warnings when using perl-style regexp:\n| The RE2 syntax error is: invalid perl operator: (?!\nfor\n| branches: ^(?!stable)\n\nThis commit fixes it by using the correct RE2 syntax\n\nChange-Id: I7efbce549a3ee673431018635c2f568f46df0449\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/35a66e3d7ee02cefa7cf80202a39dd4b9f95c9a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/35a66e3d7ee02cefa7cf80202a39dd4b9f95c9a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
