)]}'
{"id":"openstack%2Fdesignate~900438","triplet_id":"openstack%2Fdesignate~master~I0c4fd340dd657613ec69109fccd51bda0b3ec3d0","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-17 01:05:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c4fd340dd657613ec69109fccd51bda0b3ec3d0","subject":"Require correct project context for quota operations","status":"MERGED","created":"2023-11-08 16:32:55.000000000","updated":"2023-11-17 01:06:58.000000000","submitted":"2023-11-17 01:05:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900438","meta_rev_id":"3f0014c0f32202cafee8d9349ed8ac90021f5e0d","_number":900438,"virtual_id_number":900438,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 01:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-16 23:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-08 23:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-11-16 23:52:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-08 17:44:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 23:32:38.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2023-11-16 23:52:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"520916182c7c3f501510081504636ca7f2195514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-08 16:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"49a5049ae01f11fb01e1f72a4a007ab33ff95987","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-08 16:35:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7465f7a758d86b3f9533e7276ffd82bcf47dbc92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 17:44:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c0b6595c9594492a11c00ad74a20eb7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/88192309187044238ff55644f95d0ba3 : SUCCESS in 47m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a33137e2f38c444cb51cdd7c654508fa : SUCCESS in 55m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/13b1106c248e4cf085921088c5ab097a : SUCCESS in 57m 23s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/89e6eee6eb3b429d96d074fb10c1efc5 : SUCCESS in 39m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7231229784fc49b2ba90b4ee08e2d2f6 : SUCCESS in 45m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba1b4d3b0620468e98ea2821b412bf8d : SUCCESS in 1h 01m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/999f60d190e74475b10520603a1b1b71 : SUCCESS in 1h 03m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/84a78b4f802448b1b8e77fea85a47b14 : SUCCESS in 44m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/637e0e2a0f7b445cabe990fd24f5e5a7 : SUCCESS in 40m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af825bd1d71f4ca1a9f811407ebb07d4 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fb7ee2c770f4e85ae6885a8aa3b1c17 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a107636b2b845d894dd31085704206e : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d360b9ad72ed4c8689d8c00edec2141a : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f87637f60340db90f584ba813d24cc : SUCCESS in 8m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cfd25dc8def0497c9fe8f50cf365163b : SUCCESS in 38m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/826dc2c0f3fb481893454703f7f912b5 : FAILURE in 6m 16s (non-voting)\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8d324d72d055424ebb3064f5f6255f02 : SUCCESS in 38m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d4f5139198c24063a60e2aa35b2db127 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c067e6f62884856b91e1f383458cd081a62c3f5d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-08 23:32:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2b9bb2ac966e808319a7b68ddea246618bfab1eb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-16 23:52:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9c1f069c32805e0707ac9af9426ef3b84ed422c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 23:52:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb1e96567223c45e1be4e6e23ca448460936cecd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:05:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c54f0f65d984ada9b2267ac0e55f1a7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f0766ce59d824872a6405b3c41668469 : SUCCESS in 38m 43s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9946f46ab7ae4e13a94e288922d45b9e : SUCCESS in 51m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5da420aab67c43509213692ee76e9cc6 : SUCCESS in 52m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/220d5488ba2a4abdba3da41983058e1c : SUCCESS in 1h 08m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1d6060aa8f6a4d31b9a847dfa2967e35 : SUCCESS in 45m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e65ec9cfe62442219bf81ffec31e9781 : SUCCESS in 36m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e50c390a84bd4bf4b8b612faf3906826 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc2f7bd8b7304503a736d5b366268317 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3002efe4fb54156a903cd7cbd18e32a : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edcc612200074b43b61b612018953574 : SUCCESS in 8m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6a678d39b4904308b4172e06af5cccd4 : SUCCESS in 33m 10s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/df0ffa06286b40bf833345fc3e4dbe5a : SUCCESS in 6m 10s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3dbe34e8d43148f481e3b0ad903f3289 : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":2},{"id":"8f19a9ec180f32f0d8f0b333266674b4c4874173","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:05:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3f0014c0f32202cafee8d9349ed8ac90021f5e0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:06:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b7b932385d4e01b033b5cd036c2bd5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb35e27edc94740be3ab1a1f7b48789 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e7a16033141076b55f76c55e368e1cf0ec644dc4","revisions":{"9b3c16a171c3339d54fff6e4d030cc5e9b58074d":{"kind":"REWORK","_number":1,"created":"2023-11-08 16:32:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/900438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/900438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/900438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/900438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/900438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/900438/1"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-08 16:30:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-08 16:32:40.000000000","tz":540},"subject":"Require correct project context for quota operations","message":"Require correct project context for quota operations\n\nBecause of the recent change in SRBAC design, now quota apis require\nproject sopce and no longer supports system scope. So we should revert\nthese logics to bypass project scope check to support system scope.\n\n[1] c2e51939b43afcdc20e24f86595655bf588a81a0\n\nChange-Id: I0c4fd340dd657613ec69109fccd51bda0b3ec3d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b3c16a171c3339d54fff6e4d030cc5e9b58074d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9b3c16a171c3339d54fff6e4d030cc5e9b58074d"}]},"branch":"refs/heads/master"},"e7a16033141076b55f76c55e368e1cf0ec644dc4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-08 16:35:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/900438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/900438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/900438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/900438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/900438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/900438/2"}}},"commit":{"parents":[{"commit":"1912f7015ec3070849742dfaabfb12b170ef9c06","subject":"Merge \"Use super().setUp() for APIv2AdapterTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1912f7015ec3070849742dfaabfb12b170ef9c06"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-08 16:30:44.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-08 16:34:56.000000000","tz":540},"subject":"Require correct project context for quota operations","message":"Require correct project context for quota operations\n\nBecause of the recent change in SRBAC design, now quota apis require\nproject sopce and no longer supports system scope[1]. So we should\nrevert these logics to bypass project scope check, which was added to\nsupport system scope[2].\n\n[1] c2e51939b43afcdc20e24f86595655bf588a81a0\n[2] 5f87d207b448ec4e24988af5671c3b3593b13804\n\nChange-Id: I0c4fd340dd657613ec69109fccd51bda0b3ec3d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7a16033141076b55f76c55e368e1cf0ec644dc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7a16033141076b55f76c55e368e1cf0ec644dc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
