)]}'
{"id":"openstack%2Fdesignate~901080","triplet_id":"openstack%2Fdesignate~stable%2F2023.2~Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a","project":"openstack/designate","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-21 05:00:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a","subject":"Fix bug with new zone default values being ignored","status":"MERGED","created":"2023-11-17 11:30:28.000000000","updated":"2023-11-21 05:01:12.000000000","submitted":"2023-11-21 05:00:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901080","cherry_pick_of_change":900106,"cherry_pick_of_patch_set":4,"meta_rev_id":"394f9896176e72e07d1b119a932ab6910e78447d","_number":901080,"virtual_id_number":901080,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-21 05:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-17 11:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-11-21 04:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-11-21 04:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-17 11:41:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 11:41:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2023-11-17 12:51:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 04:02:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1929d9bedee2610bf9c2442734b6a8049e762a02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 11:30:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c86fd8f730b937574653f15be7b7ba41d6a78e91","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-17 11:41:24.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e8e2f64ae2be7cbdd506e886eabbc3490047fc0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 12:51:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99cc42611e404f0d9b439f447c861926\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d519316bf6f4b6ba52b10a6efcc47a2 : SUCCESS in 49m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f445888b29e04a3a818ccc44438deab0 : SUCCESS in 1h 04m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/44111e25786640f699ad86f932fad9b8 : SUCCESS in 1h 00m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/80a39822dff242fa91d44fb733c19935 : SUCCESS in 54m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2160b88095f54a64bcbbeda4067ebc47 : SUCCESS in 57m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e339daca6dc94fa8bd70f5df2d3d9df3 : SUCCESS in 1h 14m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bbc34184300145d696a074bcdb3e229d : SUCCESS in 1h 13m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1294aada7a224342989e08fe30682ffd : SUCCESS in 46m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fcb194b8ab18492eb2e00706b653591e : SUCCESS in 50m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa0e28834885439a866f3d6efc14c2e8 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1b46598470493abfd46db9956b0faa : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3cb83a7375c48e597c05b3c05c4e2a7 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4a402e6bd92429197d8ddbc5b75852e : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05e8d6a0ad994e458f93185a1dca45d6 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/803d4cd06dbe4cb4b333dc81068bf1d0 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80cad1560af410eb16fd401556644cd : SUCCESS in 8m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/70cf24c78741410d905db1d691b1c574 : SUCCESS in 39m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e9963e38fb8a4518b45cf54a0a0efd35 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53e98f66bfa91d9e2d41241e32d72e7674d193c3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-21 04:02:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bad697578edc1e2fc6f85db10dc1b872a97e479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 04:02:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef870bcf90ae8de8ff30686f9bc81adcf45232bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 05:00:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23431ae3bb9e4b78876cc12a3f47ad57\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3b78f5655ee64640a5affc3d76a13b0e : SUCCESS in 36m 14s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a969e17d0c93468d938c0c739e1c8ef4 : SUCCESS in 50m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5f675f37bf464f85b4b12c6f4e31a810 : SUCCESS in 36m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0ca831a77d44317a5859a3e47cf0445 : SUCCESS in 56m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1234ef84b3dc4ac8aa41cf44e64cb081 : SUCCESS in 50m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/204037bd80864447ae317973475d8620 : SUCCESS in 39m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a383de4b6ccd4641873d250e2a4b7288 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dd0877477564cd9b2f86f44dd499aad : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8a9a1eef694d59bb5ef2b7afcf9858 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3882053a952440bc9534de18a49a87c8 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b716644db0d94788957e40e7576e79a7 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8cde2ea2c0ab491198662048956dcc2f : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ea917b9022555158ea33ab66f7208efceaca1c07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 05:00:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"394f9896176e72e07d1b119a932ab6910e78447d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 05:01:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/232851c489fb43a8b3f824e9c8d57614\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4118f43e0d564edfbc49fc6d7e564461 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3de4f7f24b02fb77d20846daf7a7c9b9c56e1dd0","revisions":{"3de4f7f24b02fb77d20846daf7a7c9b9c56e1dd0":{"kind":"REWORK","_number":1,"created":"2023-11-17 11:30:28.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/80/901080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/901080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/901080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/901080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/901080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/901080/1"}}},"commit":{"parents":[{"commit":"53cfe80ca8b8acbb666d41c28381ee91f4dd74e2","subject":"Merge \"Fix AXFR loop when updating secondary zone\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/53cfe80ca8b8acbb666d41c28381ee91f4dd74e2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-07 14:08:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 11:30:28.000000000","tz":0},"subject":"Fix bug with new zone default values being ignored","message":"Fix bug with new zone default values being ignored\n\nThe following defaults weren\u0027t being applied properly\nto the zone.\n- default_ttl\n- default_soa_refresh_min\n- default_soa_retry\n- default_soa_expire\n- default_soa_minimum\n\nCloses-bug: #2042944\nChange-Id: Iae4ee61e4a1f2bab48b1574ea55c90d45f5a040a\n(cherry picked from commit aae83d09b88c90a5f0872daa206dbf5921aa1827)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3de4f7f24b02fb77d20846daf7a7c9b9c56e1dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3de4f7f24b02fb77d20846daf7a7c9b9c56e1dd0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
