)]}'
{"id":"openstack%2Fdesignate~901135","triplet_id":"openstack%2Fdesignate~master~I7afb533bd2d19dfaa900e3b946dc1e3563cc5198","project":"openstack/designate","branch":"master","topic":"bug-2043701","attention_set":{},"removed_from_attention_set":{"6681":{"account":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"last_update":"2023-11-19 02:43:28.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-11-17 16:21:55.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I7afb533bd2d19dfaa900e3b946dc1e3563cc5198","subject":"Add charset and engine parameters to alembic create tables","status":"MERGED","created":"2023-11-16 13:06:32.000000000","updated":"2023-11-19 02:45:22.000000000","submitted":"2023-11-19 02:43:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901135-bug-2043701","meta_rev_id":"d452dcc8db0048e7086b878468e68efb0461c003","_number":901135,"virtual_id_number":901135,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-19 02:43:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-17 16:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-11-17 16:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-18 17:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 13:06:53.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-16 14:41:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 14:46:16.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2023-11-17 16:37:59.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"810c3c2f893bb9cb6aad51bc0e4c821c94a07c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 13:06:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e982fa35f8d008a697c198fcab04224500453e95","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 13:11:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"96e222cef892f73f379654ee8cf85ef49ef16265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 13:51:34.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2079443dc0280f11c6fb122b785301b426d962a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 14:41:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b3a7f87a3e8434b8b3ec6a4f55b3f23\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9deb9bf108f4fbd84d3ddb54d8f6503 : FAILURE in 35m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/68b4785c6c1d457cbc92bbc0c179e1ff : FAILURE in 44m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aa95b9dde9d400c85ce53450c992cae : FAILURE in 19m 08s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3c974b2a009b44319b44685c6d72c985 : FAILURE in 29m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0a9b416f40014bb899104316c5f3213d : FAILURE in 29m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/566ef77dab5f434ba2bebb66f8322ade : FAILURE in 33m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb19b703c1114e2ba83904fdddf54471 : FAILURE in 31m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/11b06408404d45978ecf754b2279a521 : FAILURE in 19m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9df70aaadd88445794c84f38d19820e3 : FAILURE in 29m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5caad4f7500248ff92899df234a1b8a6 : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e22e9d729b1c45b1a20cf3cb0a75e4d1 : FAILURE in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e90303bee1e451d8047214ecd0b7556 : FAILURE in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4448eb679f64471bb383d705ffca8aa : FAILURE in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3257e0d33ddd49feaf497cf1f87b0272 : SUCCESS in 9m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/422c606d3f8c4e03b710c2754295fe7d : FAILURE in 32m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bfb9da5d553a4566849fdfea3b5f522a : FAILURE in 6m 34s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1ab4d18d8ce046ecb37103cb5a06f79c : FAILURE in 28m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7bb7dbdf671b479e93414ab468eb36b8 : FAILURE in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"400b2c114e11eb6e8197dd113fe268ea47143574","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 14:48:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"41f41bdf0f47c8e3388daf730cdbab45c1403a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 15:45:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba474bd220874611a6aa160acb28dbfc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/380ed6684db84b1c9b115c575f664485 : FAILURE in 21m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e3d79696717e4958aa7900881c7f1b33 : FAILURE in 44m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0bf2572641784debb7a1fa203904424a : FAILURE in 36m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5ad611667d464040ba7fbdd26779e1a8 : FAILURE in 30m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2d5c505cd43949f8a68f9a687917a70a : FAILURE in 28m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/aa68bfe11d9b4ff0afbf4d56279b984e : FAILURE in 50m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9aa96f55c13e44f2964cf9880ef629dd : FAILURE in 46m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/10fce3009b9a4c95ad485f0e69e23ffb : FAILURE in 33m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/92679da2cfb54d758921dd3861530d1b : FAILURE in 31m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68f492f8201a4f82a7f86844b7def2d2 : FAILURE in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74c232ec3114f0ea6c88b5cf2fc4695 : FAILURE in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fe488f3c3924ef2a250a3b9a4c8e2e9 : FAILURE in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/734674bf1f7946a69215792f853c7de0 : FAILURE in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553799308d8f4fc0815186fc7e1f69c7 : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d849916d06df4a7f92526b60118b0de2 : FAILURE in 32m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2cd69bfd5bf5437290cda3f796466f8a : FAILURE in 7m 29s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d3bf174fce1c4ed59c98119fa8999db7 : FAILURE in 31m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/659a89c8fdc74df39d6fa4aad6bf0934 : FAILURE in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d65f60eb549fb1a7d57ec7535def819da47a79df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 17:49:32.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4999126b3a39bdf99a13d01e44de13c00599e11c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-16 17:53:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d57675255c2aed2caf4067b08d13aae5c6e72fc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 18:56:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78442497fc5e416f82783bb0c9c59aea\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3aebc0f0d11343f1b9f9eb033217d18f : SUCCESS in 52m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/56d0ee903e954252970ba5170a58abb7 : SUCCESS in 1h 00m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5d38fd9ad39646938f7a1c941c0fc478 : SUCCESS in 57m 27s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7ffb0cb179c242f387df698040ddbb51 : SUCCESS in 46m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f23d50aa13404bbd8ba1d24f341fb01d : SUCCESS in 50m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/97cf36edbc78480b86dfca4f94163aca : SUCCESS in 57m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/896164f9afa34fc5bce74faab8d92b07 : SUCCESS in 1h 02m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9fd2cc32e9d4442fb3d44930a4cdadb5 : SUCCESS in 53m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a361552a10a242418fc3ad7142659b1b : SUCCESS in 34m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c6807cd2074aaf8c55038a5e97eef9 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf88d13a69044cadadd3513a7475dcfb : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57f93ac3e23f47d2b725e8166acf4827 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d689b8f6da13477d80ebcae4c3e69425 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f498997436234f9695005b630f5a7243 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a42304a3eced4163939a74d1bfd265ad : SUCCESS in 41m 40s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/342d968a3dd44fcba7b3be606100f820 : SUCCESS in 6m 49s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2e86c873fa984760bd447e3cca5f6bb5 : SUCCESS in 33m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/269b2fbf7fe54623bd47b3a6da11e410 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"37e5af975c0ff4a32c0d40fa81bde6fed20250a7","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 19:28:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d99fde3ab90e7b1e44f9db669e6741f3f4b1f232","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-16 19:42:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b7db2322c7d7fb5e992fab5ec9b40425a5751df4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-16 21:46:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"951f857c336e622728c65549c19d60b6d0d0d30b","tag":"autogenerated:gerrit:restore","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-17 00:13:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"d444074a677587ac7e15bb725feb1800af07a4e6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-17 00:19:55.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3089b82380da2d6db6fa51ba664ad071b76bd6bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 01:30:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcd62d9dcb1740f09fe46de6a5c0a9c8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6e52ed4d5184898aa56bdef75c9e3a5 : SUCCESS in 32m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/99329bd5eba84956aa1960aeb9393674 : SUCCESS in 47m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9661d6d5f6f347dfb5696fac7bf60d1e : SUCCESS in 41m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a74f3c92b7f24057941e6e2e60706048 : SUCCESS in 34m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b87db910494e495195c3121fb3f51bd7 : SUCCESS in 47m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0c67d87fd8874d97abe43e152d974be5 : FAILURE in 44m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/722beb600069420ebc91aeb090c5e57a : SUCCESS in 1h 10m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6b73c90c06d548bab6ad7adef4b69254 : SUCCESS in 35m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e2acc7f55d404ed39f95235378ef0039 : SUCCESS in 48m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b816645e004f008901de0d81dc2a94 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15015636aba54e0b92338b8715567b00 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9fa2beb5d724c3b8cd41943ffc70cea : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f64d68297474c40a2836ca65a3aadc8 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e107c9d1f94dda9f75f1cc9ebce3e8 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f96166d81beb43a591fa92d8c492ea48 : SUCCESS in 34m 21s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2bd8c9ad1e5b425d80ad65f588e83ac7 : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2a9962235f1e426290c3b6c1585c744e : SUCCESS in 50m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f088fa93051548c9837f7522c8bc8dca : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f682c07850a2f1e6de49e6291c57121211238324","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-17 12:45:52.000000000","message":"Patch Set 4: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"71d898dea87ea35f6b36ba053153d99ce3f6e3c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-17 13:11:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"031b0babdf8111818b1bd4bab77b05773ccc5072","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-17 13:13:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"11dffc287257ebd6392c4946140497f952e8d96f","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 14:46:16.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3142f36a2de7432c079d23078b875482cefdf6ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 15:04:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1fafe80a85449f89497e5602c8c349\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/caff14d500864bfdb6c9ef24f8ec1ad9 : SUCCESS in 1h 08m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2d1ec90905204a4287451e5c819b71af : TIMED_OUT in 1h 15m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/74dbb482c01c4dc581ee96b03c850112 : SUCCESS in 46m 00s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e97c4338ad3f424d82c46040ba287fd4 : SUCCESS in 38m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a297b206915447b8ba1578f83a790603 : SUCCESS in 51m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a6cbc0a841044ed6a3a6989f5b34888d : SUCCESS in 57m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/14eaf6bdc04c4ed3870c0fafc009f06f : SUCCESS in 54m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/afaba0b07f3242dba5ab3a9685374678 : SUCCESS in 36m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5f126294db5646aeab39f18cc7b5789c : SUCCESS in 57m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0eadf6887e59498c8307f929c3e8d767 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3917b594e1480ea9b4e3b9c6d7a2f3 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b265dfc5878a4e959520edf15b292348 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89e0809cdef04c4d81973a4e94c7b6a4 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ba9de43eea4211b45300feb2fdcceb : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc9824cc80f247c5bb664417eff1f984 : SUCCESS in 42m 29s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f0deeea84b5243f099ec9a1e21957cea : SUCCESS in 6m 16s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bc87cea970b74ca3a881c9c4610c976f : SUCCESS in 51m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3de934f0d189498db227d9e39bbfb0b3 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"01b971cd9788680716c96c87024facdaeec4846d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2023-11-17 15:56:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"498855558e62c7cad82e465287bb74257376a00c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-17 16:21:55.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e59d86d889b07605f9385bc88ff33ad017a8bd55","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 16:37:59.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"680c29897d904dc5242e59e42d1160bac5c61478","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:45:10.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"606bcc90b4dce33bbf86aabe3e01adccd5dac443","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-19 00:30:41.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"feecc38afb7ad8daa95c0ec644901c0f917f7b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 01:45:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6177aa23de246f7b09c489cc5cf5e45\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/26eca1ec4b2442bdbdeac36e35eb3726 : SUCCESS in 32m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/789810e1b650447daf00abcd1fd48856 : SUCCESS in 1h 03m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f45847b64c3142f68dbafd4dcdf0fa4a : SUCCESS in 31m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1ee447ee116d4601af639f18e322ea26 : SUCCESS in 29m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/73370d03d32249e5b446f202258dd567 : SUCCESS in 47m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/854fa281fb9b4f00bd81c2a720e149f8 : SUCCESS in 1h 08m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/23d82b4af1cf44eab845f8a6ce209e8f : SUCCESS in 45m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2b29d35ced0b432aac63e78454b9564e : SUCCESS in 50m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3ed434587d2a47b083438185f6e91178 : SUCCESS in 50m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6aec1c8e8a5749878c33f4fb1dde283a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b45d2f3b13c041e4ba6a4362b26dd13d : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30867d5ae02e43228db1e84dcf008e27 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fecae1c644c4653ab7fac60db024d21 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/062683f2adf9453899a763104f1e8f6a : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25fb2d388e784d32ad6fd1b2d8ba01ac : SUCCESS in 3m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4e3ab51edbc45bcb4a5d0ad2b905d19 : SUCCESS in 32m 51s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5e1efc6617384bb0bd6b3a8f88657321 : SUCCESS in 6m 46s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/daf675a129e14b34948bc1b4582b49fb : SUCCESS in 32m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dd7ac1ec4b6443b59c4310cc0bf38103 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8293bbc6b14b7b5544dfd76d2854bba34dff1c01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 01:45:53.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4e37825e01e5038895e07e8b68df593fd390e425","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 02:43:28.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deef91ab20244d0998c187f43636ee2b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/990cc6ab084a4633aa086ce7bb3b70ea : SUCCESS in 41m 39s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/09a759d7098a48528179710101c506af : SUCCESS in 42m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f181938a23be4a328ebd1ca840f09c3f : SUCCESS in 38m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a9ddcbf684f4a12a8c1fffe1e806cef : SUCCESS in 48m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bca8c1b641bb427f98f6349e1de635ff : SUCCESS in 51m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/17e0d89afcf7429ca3e9969e6bdd8cb4 : SUCCESS in 45m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874b1421046b4d5eb8a3274da8147980 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecee5676e2c54310bec3f843a132fc36 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/458e5eaabd8d485b9e5a62333f2001c0 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6701e7fb2e88495d906ddaa7e8eecfac : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cfa84984263042f38bb88d36dd0bc869 : SUCCESS in 5m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/78b77e9b016449348ffdcafea793ccb6 : SUCCESS in 32m 46s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/385194a6844049fdbbd643df11303ab7 : SUCCESS in 7m 04s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/808e41143e04463cab21a67ed5273393 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":6},{"id":"22fe5ef7d310efc8d5df47f4e0cbc4bff06fa10a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 02:43:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d452dcc8db0048e7086b878468e68efb0461c003","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 02:45:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/236de913a06f4e46967d84e672d73f30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4be3ced938ae404c94bc86a373ad586b : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f32454c31c947809d0250c2ca14de91 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1f045ee64a27c6208b3dc2a6bf5b9c5027342235","revisions":{"a595ea10d757f21d56ade712a160d5953357c6bd":{"kind":"REWORK","_number":1,"created":"2023-11-16 13:06:32.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/901135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/901135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/901135/1"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:03:53.000000000","tz":-210},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:06:31.000000000","tz":-210},"subject":"Add charset and engine parameters to alembic create tables","message":"Add charset and engine parameters to alembic create tables\n\nThis patch enforces utf8 and InnoDB when creating tables in\nmariadb/mysql.\n\nChange-Id: I7afb533bd2d19dfaa900e3b946dc1e3563cc5198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a595ea10d757f21d56ade712a160d5953357c6bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a595ea10d757f21d56ade712a160d5953357c6bd"}]},"branch":"refs/heads/master"},"614cca4eb9c114ebcde16e314fe85d0fc6648501":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-16 13:51:34.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/901135/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/901135/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/901135/2"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:03:53.000000000","tz":-210},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:50:57.000000000","tz":-210},"subject":"Add charset and engine parameters to alembic create tables","message":"Add charset and engine parameters to alembic create tables\n\nThis patch enforces utf8 and InnoDB when creating tables in\nmariadb/mysql.\n\nCloses-Bug: #2043701\n\nChange-Id: I7afb533bd2d19dfaa900e3b946dc1e3563cc5198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/614cca4eb9c114ebcde16e314fe85d0fc6648501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/614cca4eb9c114ebcde16e314fe85d0fc6648501"}]},"branch":"refs/heads/master"},"8b5e36e3e66948b0febeec20a9aa93ecef133539":{"kind":"REWORK","_number":3,"created":"2023-11-16 14:48:32.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/901135/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/901135/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/901135/3"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:03:53.000000000","tz":-210},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 14:47:38.000000000","tz":-210},"subject":"Add charset and engine parameters to alembic create tables","message":"Add charset and engine parameters to alembic create tables\n\nThis patch enforces utf8 and InnoDB when creating tables in\nmariadb/mysql.\n\nCloses-Bug: #2043701\n\nChange-Id: I7afb533bd2d19dfaa900e3b946dc1e3563cc5198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b5e36e3e66948b0febeec20a9aa93ecef133539"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b5e36e3e66948b0febeec20a9aa93ecef133539"}]},"branch":"refs/heads/master"},"5d01c3f4bb0266379072a8a8232f2cf3d9d3c6dc":{"kind":"REWORK","_number":4,"created":"2023-11-16 17:49:32.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/901135/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/901135/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/901135/4"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:03:53.000000000","tz":-210},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 17:49:01.000000000","tz":-210},"subject":"Add charset and engine parameters to alembic create tables","message":"Add charset and engine parameters to alembic create tables\n\nThis patch enforces utf8 and InnoDB when creating tables in\nmariadb/mysql.\n\nCloses-Bug: #2043701\n\nChange-Id: I7afb533bd2d19dfaa900e3b946dc1e3563cc5198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d01c3f4bb0266379072a8a8232f2cf3d9d3c6dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d01c3f4bb0266379072a8a8232f2cf3d9d3c6dc"}]},"branch":"refs/heads/master"},"314271a616d19302c22535fd6e38922d7dd12f3f":{"kind":"REWORK","_number":5,"created":"2023-11-17 13:11:02.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/901135/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/901135/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/901135/5"}}},"commit":{"parents":[{"commit":"4e185d4e266299ce22376024977cd9471dcaa0ed","subject":"Merge \"Make sqlalchemy-2x tests voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e185d4e266299ce22376024977cd9471dcaa0ed"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:03:53.000000000","tz":-210},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-17 13:11:01.000000000","tz":0},"subject":"Add charset and engine parameters to alembic create tables","message":"Add charset and engine parameters to alembic create tables\n\nThis patch enforces utf8 and InnoDB when creating tables in\nmariadb/mysql.\n\nCloses-Bug: #2043701\n\nChange-Id: I7afb533bd2d19dfaa900e3b946dc1e3563cc5198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/314271a616d19302c22535fd6e38922d7dd12f3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/314271a616d19302c22535fd6e38922d7dd12f3f"}]},"branch":"refs/heads/master"},"1f045ee64a27c6208b3dc2a6bf5b9c5027342235":{"kind":"REWORK","_number":6,"created":"2023-11-17 15:56:42.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/35/901135/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/35/901135/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/35/901135/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/35/901135/6"}}},"commit":{"parents":[{"commit":"4e185d4e266299ce22376024977cd9471dcaa0ed","subject":"Merge \"Make sqlalchemy-2x tests voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4e185d4e266299ce22376024977cd9471dcaa0ed"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-16 13:03:53.000000000","tz":-210},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2023-11-17 15:56:28.000000000","tz":0},"subject":"Add charset and engine parameters to alembic create tables","message":"Add charset and engine parameters to alembic create tables\n\nThis patch enforces utf8 and InnoDB when creating tables in\nmariadb/mysql.\n\nCloses-Bug: #2043701\n\nChange-Id: I7afb533bd2d19dfaa900e3b946dc1e3563cc5198\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f045ee64a27c6208b3dc2a6bf5b9c5027342235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f045ee64a27c6208b3dc2a6bf5b9c5027342235"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
