)]}'
{"id":"openstack%2Fdesignate~901143","triplet_id":"openstack%2Fdesignate~master~Ic977f8c3986a0c24d43db1f3899a508620096c03","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-17 11:35:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-17 12:50:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic977f8c3986a0c24d43db1f3899a508620096c03","subject":"Add basic storage test to test config overrides","status":"MERGED","created":"2023-11-16 15:59:38.000000000","updated":"2023-11-17 12:51:31.000000000","submitted":"2023-11-17 12:50:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901143","meta_rev_id":"9ce74c89fd2d12bebfb653abcf05e89715afb1db","_number":901143,"virtual_id_number":901143,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 12:50:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-17 11:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2023-11-17 02:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-16 23:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-17 11:35:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-16 17:09:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-16 18:04:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-16 23:49:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-17 02:35:06.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-17 11:35:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"bebe74826ebf447fd32c3882de565e9bf7745cd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-16 15:59:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0d724d00c1e054c9b8675af923bf7eb4469b191","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 17:09:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33513b5a417143ccb5776a27c6b69286\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9957b9413b304f4a9773931623b0923e : SUCCESS in 49m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/59bb35ce1bec4702b9752646298b7ec3 : SUCCESS in 1h 03m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/348c1bca9fdb4bbf9fa090c6bec97bb3 : SUCCESS in 1h 00m 11s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/63fd9042cf4a4940b3bc1966f090332b : SUCCESS in 47m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee3b612428584526afff54ea9f5a7e70 : SUCCESS in 30m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a7f819e1f7154ada9afee8c61cde6421 : SUCCESS in 52m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7efd5a4ad870400ca127abccbe5d98dc : SUCCESS in 56m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a4937bc476014545bd84694d9eb34245 : SUCCESS in 32m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e4eb30d53e53475686a5ef6d81eb9f2c : SUCCESS in 37m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6c0248babf44f25a07e1ad9e5cd14c6 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c63905be684dd9aa7f4658c4adb24e : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f1c6fd1db024d378f9862b5be278d1c : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f76f5e432007485e89aa54f09ad372bc : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fece9472d1c24985bf1bc4d2f438d253 : SUCCESS in 7m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eb146325f11d462292c8e78a7a379569 : SUCCESS in 39m 07s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b0e4d396c16d473cbba7fabf5fd896fe : SUCCESS in 7m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2b496d88d0d04781b0fbe3abdcf685f3 : SUCCESS in 50m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/709b3361fa84439fa642a42ebe083e12 : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d79245fa776fec50b6f6de09192d47aad8bb1769","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-16 18:04:53.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cb675aeba5cd7444bafc3671061056bc6d2a644a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-16 19:20:07.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"09884a9f13d7249922940b4f2527649599c44834","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-16 19:20:29.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0c5935214f7ad2576a4f525d65919911b897358d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 20:33:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/21733390e99e441c929b7b1f5106a07b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/18d29f912c474f47b6b0e305a4a335dd : SUCCESS in 30m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a04107d0eaee4925b1f8031fdbe99d87 : SUCCESS in 47m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/de738becfacf41beb3d4aad088c9bbac : SUCCESS in 59m 41s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7350861cbd924fbebecea594ab942471 : SUCCESS in 38m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b60c05dfc2794cfeaba63ed807d44db7 : SUCCESS in 37m 13s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/db82905fad1a4af6a67f1a3870d70462 : SUCCESS in 1h 07m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fbdac8139f534b028c8136b0b5e5f842 : SUCCESS in 1h 00m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f9627cf79404d78adbdaf817604528f : SUCCESS in 36m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/34a5212eba084e7584edbbde432bbf2a : SUCCESS in 36m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9779e58e28b54b659a68be2c0e9cd112 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f9379063a7c4998b9b621ef28a7ebd7 : FAILURE in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a6b0ce939244360816b9ade36c83927 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9bf82e6df6b4918892a5c1357d0c491 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7739a6ec8cb4410a41b73cd7ca750b1 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d4c5906d1bf34429bc19e669b10c7553 : SUCCESS in 33m 50s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/376e1416510d467da507ca01919dd603 : SUCCESS in 6m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1171d48d5b8841d2a597f3141bde4cc0 : SUCCESS in 38m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/11fa298d58c74b1ab55dd298eca4d0c5 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"318e9ec38015ed9b11e8f94003adf811d5affca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-16 21:33:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"04a56daccba4bc482fe38563886e9e4b51bfc54a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-16 21:39:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"14f0939547ac2dd84583d889be8d021e9c397abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 22:58:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10aa7cc09f3248eea5022e34231aa567\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4402391a969140f2a6bfea51035fc337 : SUCCESS in 37m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/25b536c2e78542158a657d440d78120a : SUCCESS in 48m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a8a356a774e408aa01b382796b2f69b : SUCCESS in 46m 53s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/da9eace0e8294a79a69697bf86fd8cc9 : SUCCESS in 40m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9893d9eea32f4c6795ec2450687a0807 : SUCCESS in 39m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8bd5f95db09540dbb31c7833721a5e2b : SUCCESS in 51m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/76ad8e7c49b54a119f000089116519aa : SUCCESS in 1h 11m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e47a469e729470aa94fb882a0a47156 : SUCCESS in 47m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f704066aeac4f089efd608f770f0494 : SUCCESS in 32m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8559016d7ee8409cadac3a9dd0bbdc07 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86942805ad0a43ef9ecab7a662214be3 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42a3eec4145d4734a9e8b27704920804 : SUCCESS in 6m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8d791b9d17e45f4a2b69fe77e2b381d : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32cb6e3653a9490899e59781d1c730a9 : SUCCESS in 10m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e833ea427d584e0c878ce0bad52fe808 : SUCCESS in 40m 08s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f2cc39223f934c5089ead405a2acdd2c : SUCCESS in 6m 38s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ec46581b8b7e4b0ab722b3f058dc9038 : SUCCESS in 49m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/41b04a8d5d0f474286a4811d581a2059 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"90fa793423c699cb095accde52abd077b4c414ea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-16 23:49:37.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d04022fc8d59ac42a27f19aa306189b9c9afd25","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-17 02:35:06.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"215f5a0572973a05fe94cdcfb54a6b79aaaf8183","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-17 11:35:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"662b5e11f6f90c9c16f47e22e31144dc771f7ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 11:35:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1e52e92c4a02dd1589183ea2ccebdec9696f894c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 12:50:26.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/544ec4d95ebf44b68b4fae0ee6dfb51f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee2be3171bba4d0cafd6bfa44215d8a1 : SUCCESS in 40m 19s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e9fe90e968f04af0951eaaf5d038894c : SUCCESS in 40m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2010faec09ee41979311252df0921c70 : SUCCESS in 39m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4cae7e299d184c0084825739493fb936 : SUCCESS in 1h 07m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/513159df13eb4c86a293a176d61db6b4 : SUCCESS in 50m 21s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/78d3b0668716481e806ea53bd44ddcf0 : SUCCESS in 41m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6f51d9724843e5bfa0cac7b05abf0b : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa2579b6b69d41959d1f20a78bd8b0db : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9754436c58154f8092b2bd32fedccff6 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41fdda29a57a49f3af98e197c763c985 : SUCCESS in 6m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f4b4670d805b44b780f35f134556312d : SUCCESS in 42m 20s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/911ecf6da77043ccbbc6e6a549145b4f : SUCCESS in 6m 33s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/d03c8268a3a3467ca670534b5975e391 : SUCCESS in 52m 58s","accounts_in_message":[],"_revision_number":4},{"id":"affe325fd47d84cc95ddf34635d4f955022f8b6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 12:50:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"9ce74c89fd2d12bebfb653abcf05e89715afb1db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 12:51:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfcfacfed6aa4c488735bbc77628982a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc3c46b1b724fe4acc34ec2f4a07dbe : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"12362a234f74ebaf8c8f7fce954a853716f9663c","revisions":{"ac43c30d5cb22271935a4c997589ae2855643f05":{"kind":"REWORK","_number":1,"created":"2023-11-16 15:59:38.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/901143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/901143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/901143/1"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 15:58:53.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 15:58:53.000000000","tz":-480},"subject":"Add basic storage test to test config overrides","message":"Add basic storage test to test config overrides\n\nChange-Id: Ic977f8c3986a0c24d43db1f3899a508620096c03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac43c30d5cb22271935a4c997589ae2855643f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac43c30d5cb22271935a4c997589ae2855643f05"}]},"branch":"refs/heads/master"},"f18d1b4901e97b3f87fbb59669961e9a43cb6234":{"kind":"REWORK","_number":2,"created":"2023-11-16 19:20:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/901143/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/901143/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/901143/2"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 15:58:53.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 19:19:46.000000000","tz":-480},"subject":"Add basic storage test to test config overrides","message":"Add basic storage test to test config overrides\n\nChange-Id: Ic977f8c3986a0c24d43db1f3899a508620096c03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f18d1b4901e97b3f87fbb59669961e9a43cb6234"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f18d1b4901e97b3f87fbb59669961e9a43cb6234"}]},"branch":"refs/heads/master"},"a8499623fa446a70568a501ff246baa5654b7561":{"kind":"REWORK","_number":3,"created":"2023-11-16 21:33:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/901143/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/901143/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/901143/3"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 15:58:53.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 21:33:05.000000000","tz":-480},"subject":"Add basic storage test to test config overrides","message":"Add basic storage test to test config overrides\n\nChange-Id: Ic977f8c3986a0c24d43db1f3899a508620096c03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8499623fa446a70568a501ff246baa5654b7561"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a8499623fa446a70568a501ff246baa5654b7561"}]},"branch":"refs/heads/master"},"12362a234f74ebaf8c8f7fce954a853716f9663c":{"kind":"REWORK","_number":4,"created":"2023-11-16 21:39:46.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/43/901143/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/43/901143/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/43/901143/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/43/901143/4"}}},"commit":{"parents":[{"commit":"aae83d09b88c90a5f0872daa206dbf5921aa1827","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aae83d09b88c90a5f0872daa206dbf5921aa1827"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 15:58:53.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-16 21:38:45.000000000","tz":-480},"subject":"Add basic storage test to test config overrides","message":"Add basic storage test to test config overrides\n\nBasic test coverage for overriding the default_ttl for\nzone creation.\n\nChange-Id: Ic977f8c3986a0c24d43db1f3899a508620096c03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12362a234f74ebaf8c8f7fce954a853716f9663c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/12362a234f74ebaf8c8f7fce954a853716f9663c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
