)]}'
{"id":"openstack%2Fdesignate~901303","triplet_id":"openstack%2Fdesignate~master~If529f292bd62269b777b20b0d2325456c8829024","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-21 11:26:17.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"If529f292bd62269b777b20b0d2325456c8829024","subject":"Improved RPC coverage","status":"MERGED","created":"2023-11-17 20:45:16.000000000","updated":"2023-11-21 12:27:04.000000000","submitted":"2023-11-21 12:26:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901303","meta_rev_id":"be2824af55d50d5de1a4c99a94ca5ba61ec416a5","_number":901303,"virtual_id_number":901303,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-21 12:26:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-21 04:39:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-11-21 11:26:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 22:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 04:39:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"9fd3b08aca9364f677592b486d7cfee0188e0e00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 20:45:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c21d0729ab5d78b5e4a5e1cb499b2127ddc0e883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 22:05:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbc23444bfa14174a10e60c88a0f2a99\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f85ac3e746bf4d0497e0076faa3bde59 : SUCCESS in 45m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/dce7bc4035934322a74bcc61babee86a : SUCCESS in 1h 02m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1bb84c78fb72455c88bed4d54eabc42c : SUCCESS in 58m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/293c65a105114d579d58116648d66285 : SUCCESS in 48m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c510093cbe24e1980e372f3853f810b : SUCCESS in 46m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a8d4024970784868b864b095064b5101 : SUCCESS in 1h 13m 51s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/234bb38a04f843eeb19e820e530749fa : SUCCESS in 1h 14m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4345eccd789d4749853eb982796fa06b : SUCCESS in 30m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1e565ee68d7d462796aaed43af63124d : SUCCESS in 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92e5eb2bd83d4006858a3931e647ae01 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e578cfeae94b18bb7a9d2c4080456c : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e59b9a36bd034bf99cca48f9d0e25a66 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1deb7933b294f02a77924bbde115b20 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544185c53cb74f8a9ae2136ac5f1a31a : SUCCESS in 6m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d24e15f0aa92495db932ddad4406d8df : SUCCESS in 33m 55s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4836a317dc0e4313b3fc6adbf5f83971 : SUCCESS in 6m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/92ed0712491f436da5022e0e7ebb9ffb : SUCCESS in 34m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d546c1ee09cc49f7b95015d86c74ea65 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5231293653ceb3c9f3575ed0d875141e4df289b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 15:51:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b72ac4795c2f6e198eace82284030fd8fb77f86d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 16:19:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad9d95c47123034328228701bf570e75c0e96760","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 17:32:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8de91cacfefb49eb8fe36367370848de\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8b8e27190da94ecab473bb84f4f5e79a : SUCCESS in 52m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c9c9bd5e54e042b681518809b6d0ad9a : SUCCESS in 59m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecfdf93111c7404db842d5ce14543be5 : SUCCESS in 58m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a85de6c6b4794f13870018f53488d4ae : SUCCESS in 51m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/07472dc244f74f7da153e9fd9f96e6c0 : SUCCESS in 46m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eed4929df86745068f595cfcbab065ab : SUCCESS in 1h 06m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/defe8f4081b4497881f2c7940e090959 : FAILURE in 1h 07m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/857bffa33a5b44feadd14cf0bc4e10eb : SUCCESS in 28m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3ab570a943934730a9517d7d79b96974 : SUCCESS in 30m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1975a373a5694c7a85b178e104fa8101 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c846cc081c4a4a7298cb2b35924183c8 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/807445b36583434380eeb3ecb6b19b58 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ae03fb51b38445c9fe1a8bb8795f857 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0659415c1df4afcbf66e46359a5da3b : SUCCESS in 6m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb560371c7954bee80858d2a049ed106 : SUCCESS in 31m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c4ea29391de74d7ebc4370b571e9d20c : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1f1e0528c48d47bea6a34d76bfd5fd96 : SUCCESS in 30m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/caced6390b8a4e19871bc8eb881ff141 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69cd648e2df330d7328a1a8cd4ff4135c569138a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:40:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f390789af9c3404377b11185e7206de27b432dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:43:51.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9e467ffdb78b337d1942a65f915afd171a1e403e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 18:53:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79ce79aa535420098cc5d9f6aba1b4b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b9173d989c95493c9e6924c1cca342ae : SUCCESS in 40m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ce637aaa5d3e4e8588caf6ef727727a5 : SUCCESS in 50m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aeda3f093593471b8c12a7b509328a1b : SUCCESS in 36m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/828048a50d284c21b67f39d36459876d : SUCCESS in 47m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/13c44d87b72644999e68deb686bae54e : SUCCESS in 47m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9969510aad504c5c836ba822437dc493 : SUCCESS in 1h 04m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8c05117ef01e41a9adf45bfc10b36f24 : SUCCESS in 1h 00m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec55660d5170407c87af37077ca22444 : SUCCESS in 33m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a7829daf755c47dab93b314ecf5b6d79 : SUCCESS in 33m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9293b4c689fc4f238ad3f4afe58afddc : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cd8fb9a4a214bb68ef7b180d5bac803 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e862cad1d75f47dfa0e833eecadfa423 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc911e4addff4df8b54f0ee3fdf2f077 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1f7f4530cb4e0d814a54b3bcb26d35 : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d6fa9cb6e69c49b58954572db154598c : SUCCESS in 32m 14s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/436bbe910ee94cdd83e172695e2800c3 : SUCCESS in 6m 06s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c8c04b7810a74b3cb57f693c510da826 : SUCCESS in 46m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ad1fc1d47cc24235b5a52bacfb0b604c : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2303efa67227418aa6011aa2a21a0f2df7cb3d7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-19 12:30:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3563de77f80e2f8e5f13491867c9483f48f55c95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 13:50:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da851167fd543e88c79104aa309e178\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/551a15806f2f4a618eecb30aad55b0e5 : SUCCESS in 41m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f6d647ceb963427292f10ad2e92bb3f1 : SUCCESS in 47m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/84283ad498a34d0b8893d58f10482e14 : SUCCESS in 48m 05s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/512aecd5b5814ed984a63912824cb4d9 : SUCCESS in 30m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c05e040af17f4a6396a2e4385d99e7e5 : SUCCESS in 39m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c8e65ab7dc664702b5d06f965cd17eec : SUCCESS in 1h 00m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ae9f6ab75d27410592c6a886167166bb : SUCCESS in 1h 11m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c4833872dd344685a9817b9bd32e7205 : SUCCESS in 49m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7a5010ee70fc4a3b9be573ade5b7e94a : SUCCESS in 30m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea94bb26cc8c492fbe34b27f4f7da00b : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d820e046c7a54362b32544618fc007ed : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f5efdc5a484d34825fd28ad0023f23 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e03830ffbf841848c9c9f4103354dc3 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/562f5f26d6324833a1fee1d555f5bcc2 : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04597bc0f44f4def91015dd9be07ac49 : SUCCESS in 32m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/93931f48669042d186b0650c73a40189 : SUCCESS in 5m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f5a679f9faf04371a89de4fc2c7a99cb : SUCCESS in 30m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/df3dba34b4054979a480fb33e0f21f1a : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2afc049b09abc2117d89c07c80fbce78e82d25af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-21 04:39:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e0a46281103242e63025fc2a52470fc1ebe117fc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-21 11:26:17.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b8c0137297ab92928113aeb98165350eefbe4fd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 11:26:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4246fb680157e7604fc2b5b08087bf2401bd4f86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:26:04.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23f7174b878b466da300f6d02f30e2d6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/962d8129aadf4fd4a7ed816f3296d5ba : SUCCESS in 49m 10s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8f04675be8e0484aa7eaa69980f34801 : SUCCESS in 45m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/328d8737a7744dae88ba3a6c9b8424f4 : SUCCESS in 39m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1df2299bf6d74c4ab2e0a67e5c496142 : SUCCESS in 58m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a3787d9a35ef45fa8ae47b0bf2bceda5 : SUCCESS in 43m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2573692bdde44044ac39181fc8055783 : SUCCESS in 48m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0607912fdcd84cfaa62a131040a1d2cf : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/190758b54e07444598c826fb5e80f500 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2fb865d68d2b4d4cb115d214f411264d : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cb59c532674b4295cf1db29736aeda : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5d9f5f7aa5d4b608ac5a01579b793ed : SUCCESS in 41m 55s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e6e7e11b885e4cf1a26b33773cc8e461 : SUCCESS in 7m 10s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7349307b23ca4fc293bf0b653063854b : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":5},{"id":"1bee901267ac0c301b7aa5438a7f1198c19535cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:26:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"be2824af55d50d5de1a4c99a94ca5ba61ec416a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:27:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5f7555ef3340e3bbd5f85c684d2dde\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139c65057ef644fb8fb8f2029f4ed682 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"9a626ef1c0e9030b826bec43a5da5b4ee14561b6","revisions":{"8bcde2362fc642b6964edec8b1aeb3a89f317bd1":{"kind":"REWORK","_number":1,"created":"2023-11-17 20:45:16.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/901303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/901303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/901303/1"}}},"commit":{"parents":[{"commit":"8805a02bedad432faf717e4be551ff1024d92783","subject":"Merge \"Add basic storage test to test config overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8805a02bedad432faf717e4be551ff1024d92783"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 20:44:50.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 20:44:50.000000000","tz":-480},"subject":"Improve RPC coverage","message":"Improve RPC coverage\n\n- Modification to cleanup to make testing easier.\n\nChange-Id: If529f292bd62269b777b20b0d2325456c8829024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bcde2362fc642b6964edec8b1aeb3a89f317bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8bcde2362fc642b6964edec8b1aeb3a89f317bd1"}]},"branch":"refs/heads/master"},"61d8e3f46f9b803df026b11177c080ca6a742544":{"kind":"REWORK","_number":2,"created":"2023-11-18 15:51:35.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/901303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/901303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/901303/2"}}},"commit":{"parents":[{"commit":"8805a02bedad432faf717e4be551ff1024d92783","subject":"Merge \"Add basic storage test to test config overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8805a02bedad432faf717e4be551ff1024d92783"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 20:44:50.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 15:51:23.000000000","tz":-480},"subject":"Improve RPC coverage","message":"Improve RPC coverage\n\n- Modification to cleanup to make testing easier.\n\nChange-Id: If529f292bd62269b777b20b0d2325456c8829024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61d8e3f46f9b803df026b11177c080ca6a742544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61d8e3f46f9b803df026b11177c080ca6a742544"}]},"branch":"refs/heads/master"},"edb323a32df8676bc9d4b0cc47fa37de2dd0914a":{"kind":"REWORK","_number":3,"created":"2023-11-18 16:19:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/901303/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/901303/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/901303/3"}}},"commit":{"parents":[{"commit":"8805a02bedad432faf717e4be551ff1024d92783","subject":"Merge \"Add basic storage test to test config overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8805a02bedad432faf717e4be551ff1024d92783"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 20:44:50.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 16:18:55.000000000","tz":-480},"subject":"Improve RPC coverage","message":"Improve RPC coverage\n\n- Modification to cleanup to make testing easier.\n\nChange-Id: If529f292bd62269b777b20b0d2325456c8829024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edb323a32df8676bc9d4b0cc47fa37de2dd0914a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/edb323a32df8676bc9d4b0cc47fa37de2dd0914a"}]},"branch":"refs/heads/master"},"9f6818f90e459212443d8d9b0c5a9974775e1cab":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-11-18 17:43:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/901303/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/901303/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/901303/4"}}},"commit":{"parents":[{"commit":"8805a02bedad432faf717e4be551ff1024d92783","subject":"Merge \"Add basic storage test to test config overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8805a02bedad432faf717e4be551ff1024d92783"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 20:44:50.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:43:44.000000000","tz":-480},"subject":"Improved RPC coverage","message":"Improved RPC coverage\n\n- Modification to cleanup to make testing easier.\n\nChange-Id: If529f292bd62269b777b20b0d2325456c8829024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f6818f90e459212443d8d9b0c5a9974775e1cab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f6818f90e459212443d8d9b0c5a9974775e1cab"}]},"branch":"refs/heads/master"},"9a626ef1c0e9030b826bec43a5da5b4ee14561b6":{"kind":"REWORK","_number":5,"created":"2023-11-19 12:30:14.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/03/901303/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/03/901303/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/03/901303/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/03/901303/5"}}},"commit":{"parents":[{"commit":"8805a02bedad432faf717e4be551ff1024d92783","subject":"Merge \"Add basic storage test to test config overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8805a02bedad432faf717e4be551ff1024d92783"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 20:44:50.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-19 12:29:44.000000000","tz":-480},"subject":"Improved RPC coverage","message":"Improved RPC coverage\n\nChange-Id: If529f292bd62269b777b20b0d2325456c8829024\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a626ef1c0e9030b826bec43a5da5b4ee14561b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9a626ef1c0e9030b826bec43a5da5b4ee14561b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
