)]}'
{"id":"openstack%2Fdesignate~901305","triplet_id":"openstack%2Fdesignate~master~Iea2a3d6f2c873de2d8757b80249d6a72a9ae31c8","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-27 23:27:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Iea2a3d6f2c873de2d8757b80249d6a72a9ae31c8","subject":"Add context hard_delete coverage","status":"MERGED","created":"2023-11-17 21:17:06.000000000","updated":"2023-11-28 00:33:38.000000000","submitted":"2023-11-28 00:32:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901305","meta_rev_id":"c567cd2d02226ce4961ec8b69196390b3b86838e","_number":901305,"virtual_id_number":901305,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 00:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2023-11-21 06:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-21 04:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-21 10:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-11-27 23:27:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-17 22:17:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 04:21:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-21 06:30:49.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-21 10:01:55.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"cefbd83614ca4d07dee829558b463c1f2c918c64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 21:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"036ed75f71e8e76de101aa384b0f80ac49407e41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 22:17:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f16587e43b048a5af504099af55fd26\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/356dac7eae184e9e970ebcfb2cc62605 : FAILURE in 48m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c573b2fe7b984e0db0216aeccf072ca1 : SUCCESS in 46m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d4b450df6ec44c3b7cc221111d6f305 : SUCCESS in 44m 46s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6b1f98439be24dd182902df8b9dd3c8b : SUCCESS in 37m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/24f4e624d262465b9406b94677e49035 : SUCCESS in 32m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/58609bc774e0476ba71ec44d3d571186 : SUCCESS in 54m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5ec8f4fbd624a73979d411b6fc5b09b : SUCCESS in 58m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3103165b64814b39b26667529c73c976 : SUCCESS in 38m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/af0153be8a1846ed9ae42fa32c81b155 : SUCCESS in 35m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e419a63546a54125b9cc428f0cf58c28 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f664b4170b840438c137a7e5d06d39c : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa2f929f5aea4a8e973246d0fc4d8994 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f6e86b256852417fb6894493b868b764 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9d8e907dba498cb031a17b600b4511 : SUCCESS in 6m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07b3b324991b4efdac080fbcf0a11a7f : SUCCESS in 32m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0647e53defc14668a8638698fb5fa719 : SUCCESS in 7m 13s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3ec15c4dd87341e4b11f66f432cf88ee : SUCCESS in 34m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9b55a07d585a427b9add44e82e154d2f : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56da4e54f90c10d997478f30cb420d30ae2718e1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-17 22:39:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d29d9d9aab04b902f5e8a1d1a7630dcc30a2b2e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 23:47:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a22c3a3ad84bccb667a71c232fb956\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7b9517ce8d144bd6b5f1a5bded4f76f8 : SUCCESS in 32m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/62e13b17d84f49b9977ba40c890d7e28 : SUCCESS in 38m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/95b7445ec8a0427e933834b8c3f2d30d : SUCCESS in 34m 57s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f2959e51a3cd40c6bfbf5d1e089a686c : SUCCESS in 49m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d3c3273786954f8a95041fe4ad97eaf6 : SUCCESS in 54m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/64c313f5c3ce4c9f818d96ab978005fd : SUCCESS in 50m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/86cebb6535ff454f89a6fe2becd5294f : SUCCESS in 1h 06m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7e8fcfba0aa4ac78be7c7988dc66987 : SUCCESS in 47m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a75a53ac554b45518395e4582bcfb317 : SUCCESS in 46m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db4043b76eea45f694854bc4f4e73cf4 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba8ae6b4ac648f393834bec87ca75a4 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9545384daa054aae9b24d051d67d4fec : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9e8b7c8a469478e8e151ef67f01f3f3 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e00eca3c3af422faccf1a21e8e5dc29 : SUCCESS in 7m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a67bb9a31f3c40cc81ba1351a1204e46 : SUCCESS in 37m 05s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/695f411d2b6540c6832a09843a2e507f : SUCCESS in 6m 48s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3c6fe2c0c98643b0b8c2469a4e492124 : SUCCESS in 33m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8ada488ccb09413fa5329b09eb40df56 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3879b30a3e4b760ab501b94782ecd576faddb9e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-21 04:21:59.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea35c023e53b1e404f32c2801545a5efc4e86d1a","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-21 06:30:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1ee131d0efaa0470c69a17649aae5fb6cc92a88","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-11-21 10:01:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4f9021c461e596f301939a43aa601333bbb0921d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 23:27:30.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a594dc3e5d84c8d9d7c275811137bfba6211e41e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:28:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f51ecacc7a85e878b83aca0765ab073368d051d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:32:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3effce627c2458385141eb0fac5a1dd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b9b5d02c71bd45e6b701ab244a1c8507 : SUCCESS in 34m 32s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/912999f7cbd443e2b68b3d78f67506a9 : SUCCESS in 37m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2c9b8497857b4696b61e9c91019c12af : SUCCESS in 39m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f1a89a8065d4776ae81f8f2c7f78f8d : SUCCESS in 59m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be94199aad1c4dd29af9a1960ed1cfcb : SUCCESS in 51m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6a435a229fb54238b15795c2d594fcb0 : SUCCESS in 48m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430828e1f14e4baca4d5c7bd7c3b3baa : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69bad2bdc0174ceab676b8ec2ac01e33 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69093c8d4f3f41839921a890d5cd83c4 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f0afd37d984c9db2428c4bfb49a245 : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/096a53f926d24042ac1be1bd507fd4da : SUCCESS in 34m 00s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/72c7a5ffd4f4431d8a6a4d7db82035f9 : SUCCESS in 7m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/c65ebff35c504277b750ecb124bf579c : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":1},{"id":"96171014d85cbd28c143d1f2cad630b1718227ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:32:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c567cd2d02226ce4961ec8b69196390b3b86838e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:33:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b9dcf983d8a4fdc954677e651064000\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139f25732a9243cb864745a11e9122a6 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4aa3a0564ffeb5394d9652b716ca8d7bcdc69e7f","revisions":{"4aa3a0564ffeb5394d9652b716ca8d7bcdc69e7f":{"kind":"REWORK","_number":1,"created":"2023-11-17 21:17:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/05/901305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/901305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/901305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/901305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/901305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/901305/1"}}},"commit":{"parents":[{"commit":"8805a02bedad432faf717e4be551ff1024d92783","subject":"Merge \"Add basic storage test to test config overrides\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8805a02bedad432faf717e4be551ff1024d92783"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 21:16:52.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-17 21:16:52.000000000","tz":-480},"subject":"Add context hard_delete coverage","message":"Add context hard_delete coverage\n\nChange-Id: Iea2a3d6f2c873de2d8757b80249d6a72a9ae31c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4aa3a0564ffeb5394d9652b716ca8d7bcdc69e7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4aa3a0564ffeb5394d9652b716ca8d7bcdc69e7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
