)]}'
{"id":"openstack%2Fdesignate~901311","triplet_id":"openstack%2Fdesignate~master~Ic75c1e2c4b4c556dab4f3b74ebf751d54bce38ec","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-21 11:26:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ic75c1e2c4b4c556dab4f3b74ebf751d54bce38ec","subject":"Increase minimum test coverage to 95%","status":"MERGED","created":"2023-11-18 17:39:49.000000000","updated":"2023-11-21 12:49:26.000000000","submitted":"2023-11-21 12:47:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901311","meta_rev_id":"12f68b5f48fb068cebad7b891da1f705e2f4a5c0","_number":901311,"virtual_id_number":901311,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-21 12:47:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2023-11-21 06:30:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-21 04:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-21 10:02:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-11-21 11:26:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-18 18:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-21 04:41:55.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-21 06:30:14.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-21 10:02:15.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"dae5b41c415cf10702db0cf2db690d823a629f4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:39:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5b46f162c0a6ef591ea15231e177e7017b0f7dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:42:33.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0328a950899cf7ef15a7ede2b3b4b3604232c653","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:42:38.000000000","message":"Topic most removed","accounts_in_message":[],"_revision_number":2},{"id":"97e6cbe65e1dba508d9f004bb895d3cf0dbad80a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-18 17:44:05.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"3d4b45dada25d60d4bd3e9f26e1f7e0e823a7f9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-18 18:58:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/745fb15e0b3649d18e4a0c1dd8eb6589\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/725ff87c40304ed889f6cc3250a77d0a : SUCCESS in 3m 48s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b6b5c1fa6b1d4dc19934aa746df239ec : SUCCESS in 44m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e10a313a276740fdb3edf46cafb3809a : SUCCESS in 1h 01m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f454732fb0f44013b499c2e76e4e3f98 : SUCCESS in 43m 17s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cc1277f992a4413b99e7c17559b50d95 : SUCCESS in 36m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0af3dc3cfa1f40d5988557a327689c61 : SUCCESS in 44m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/196890602dce4d42b12b861cf7523280 : SUCCESS in 1h 13m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f8f8c8c1e2014452ad1ccd541f9dc2f0 : SUCCESS in 52m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3eb2ca7bfc744ba9b04f2df8b8ee9e1f : SUCCESS in 31m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/de6d101138b54ccf870e9adc1309919c : SUCCESS in 39m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18183e62b7124a888fc7e493d7d46e50 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9deef986070402eaf1089689e697bea : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53f9207d4c1945bd97be6d0e664bbefd : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e93301ddbb94463cb760067abcd501ba : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6102b23900714a14982be95758e62da9 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45ad938caa684551ba7e399d5b1f283f : SUCCESS in 4m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/74b52671920242159eda14b6bac35746 : SUCCESS in 6m 17s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9b44141de0bd40ffa91803f1d50abf8c : SUCCESS in 37m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/90df05ec50fc426b91d2bbc0c7de085d : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7feb6bcff79b09d833fa467c60bc16b67d676377","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-19 12:30:44.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"20632e46a635afed27b5ec3369d81e7960b67e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 13:48:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d5576ab17b40498181ea5b185053f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/989a4b7054eb4556ab6045f071a7a269 : SUCCESS in 5m 15s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/299c3a48ecfa4e4ab5ac4bf096e32932 : SUCCESS in 31m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0f1aa57f32564b3ab633c10c4e007fa1 : SUCCESS in 1h 01m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/610db86e4c3145b5be5c29bd3b60c616 : SUCCESS in 55m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4268c45d7dff4e5498d241c645a233ac : SUCCESS in 50m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c19c3b65fbb946cfaa20b07dfa39b4f7 : SUCCESS in 37m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/26b0a7d7a8d74126aefbeaa268d29e27 : SUCCESS in 1h 11m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/978700b434654d51b9c84f983ac7bc7e : SUCCESS in 49m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a783c84ebeae4ec38c82eda2cd6d612a : SUCCESS in 32m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/03c99a74efdf4422a47ec106372cf89e : SUCCESS in 47m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fabcb81121214dfca77b0398ef3da2e0 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604aa5d0c9da4c349a1c12ead8273d70 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbcfb650afb24f36b5be7227f65fd459 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5d086ffd1b34ef5b94635e957101e22 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f601771a427541049d14099c1487fb93 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee9fedd378df45cfa183ae26c1a3c7d9 : SUCCESS in 5m 12s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a06510c188b744d6982d283d2d9a0360 : SUCCESS in 6m 54s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/0664162687d340d18b320134b94823c1 : SUCCESS in 43m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e71c7ef914e14ec29e467f26c599d8dc : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6c8b1630c20697cef4832e50e6ad031dfda5dccb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-21 04:41:55.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"68ec2fa585e3ff8c472356ab1a50a5456189c271","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-21 06:30:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b163e470be3aeef9abd2d0ebe0ef1d808f436409","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-11-21 10:02:15.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2cc7f3da7306cc99023639cff64d102cc9bb4d63","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-21 11:26:43.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2316021d300dea7057be230b933336d413b0c769","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 11:27:35.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"65c1f5568bd01054153653899d6df0855c4285d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:47:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a898e20d926f400a8f6d0a8c60af31c4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/10f5dd434e0b48a6978cfef0ddcbffc3 : SUCCESS in 4m 51s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3332be3cb9e246daab29d2ffb7e31c39 : SUCCESS in 47m 51s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5140a42263b04928a6f5a92a35640be2 : SUCCESS in 50m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d01a698ad2bb4fa992ea179f73fefe55 : SUCCESS in 50m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/94241caa9a844abe84c4b74ed4139c20 : SUCCESS in 1h 14m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d3d270b945f469699258b0677ce21c7 : SUCCESS in 47m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8bf0f8277365433ba6cd83f01ab0b0bf : SUCCESS in 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af1911c076c54e058862f95a1560c8c1 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e87325c715f4a0b944cfc0409350be5 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8dc5b71ce924bb287c3da9529c341c0 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305ef713ec4647f69b75ca96f3bcf992 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/722a12c140bf4f2480e2505cec3630dd : SUCCESS in 4m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/270b3bbd9ba449528e6eafa24b2e2d51 : SUCCESS in 6m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6038d1549f4944c8b33e42bd83af3036 : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":4},{"id":"649fc6fd25e371e1cb75c8fe6207e086cc2dc03d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:47:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"12f68b5f48fb068cebad7b891da1f705e2f4a5c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 12:49:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0048ee2564340f9837d970294cdc5f3\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/8b25b3c87c3548e99188d8e2cf1a3d1b : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52448e01c524ff0b06872f135e19379 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f70d28305eb4acaa533cb894fd8ccf4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"27ffc68f7b7f3175c3883cacaded2a9b73c2dfbc","revisions":{"faebb2d487bc6c7addc06535b770e6e544607735":{"kind":"REWORK","_number":1,"created":"2023-11-18 17:39:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/901311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/901311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/901311/1"}}},"commit":{"parents":[{"commit":"ab925cca07ca2e66d1993e7f5a1ee2b3b44176cd","subject":"Improve coordination coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ab925cca07ca2e66d1993e7f5a1ee2b3b44176cd"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:39:24.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:39:24.000000000","tz":-480},"subject":"Increase minimum test coverage to 95%","message":"Increase minimum test coverage to 95%\n\nChange-Id: Ic75c1e2c4b4c556dab4f3b74ebf751d54bce38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/faebb2d487bc6c7addc06535b770e6e544607735"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/faebb2d487bc6c7addc06535b770e6e544607735"}]},"branch":"refs/heads/master"},"6741c593775a192ca34081b0a5c5bec88ffdf794":{"kind":"NO_CHANGE","_number":2,"created":"2023-11-18 17:42:33.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/901311/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/901311/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/901311/2"}}},"commit":{"parents":[{"commit":"fa16c2b61c87426d5a2934cb7cd988173f39f744","subject":"Improved coordination coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fa16c2b61c87426d5a2934cb7cd988173f39f744"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:39:24.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:42:33.000000000","tz":0},"subject":"Increase minimum test coverage to 95%","message":"Increase minimum test coverage to 95%\n\nChange-Id: Ic75c1e2c4b4c556dab4f3b74ebf751d54bce38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6741c593775a192ca34081b0a5c5bec88ffdf794"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6741c593775a192ca34081b0a5c5bec88ffdf794"}]},"branch":"refs/heads/master","description":"Rebase"},"2a6dec06cbb89797aa7529ee2450994917e96c64":{"kind":"NO_CHANGE","_number":3,"created":"2023-11-18 17:44:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/901311/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/901311/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/901311/3"}}},"commit":{"parents":[{"commit":"f0ec080c51e621f988508b1865dedc119253d6f2","subject":"Improved coordination coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0ec080c51e621f988508b1865dedc119253d6f2"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:39:24.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:44:05.000000000","tz":0},"subject":"Increase minimum test coverage to 95%","message":"Increase minimum test coverage to 95%\n\nChange-Id: Ic75c1e2c4b4c556dab4f3b74ebf751d54bce38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2a6dec06cbb89797aa7529ee2450994917e96c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2a6dec06cbb89797aa7529ee2450994917e96c64"}]},"branch":"refs/heads/master","description":"Rebase"},"27ffc68f7b7f3175c3883cacaded2a9b73c2dfbc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-11-19 12:30:44.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/11/901311/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/11/901311/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/11/901311/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/11/901311/4"}}},"commit":{"parents":[{"commit":"8054bba0acaa21166863304c078cf604c11a0886","subject":"Improved coordination coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8054bba0acaa21166863304c078cf604c11a0886"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-18 17:39:24.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-19 12:30:44.000000000","tz":0},"subject":"Increase minimum test coverage to 95%","message":"Increase minimum test coverage to 95%\n\nChange-Id: Ic75c1e2c4b4c556dab4f3b74ebf751d54bce38ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27ffc68f7b7f3175c3883cacaded2a9b73c2dfbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/27ffc68f7b7f3175c3883cacaded2a9b73c2dfbc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
