)]}'
{"id":"openstack%2Fdesignate~901582","triplet_id":"openstack%2Fdesignate~master~I16ee9a86f64af2d666adc8a5cad5267fc34910b1","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-12-12 03:59:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-13 00:02:21.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I16ee9a86f64af2d666adc8a5cad5267fc34910b1","subject":"Improved service coverage","status":"MERGED","created":"2023-11-21 14:31:57.000000000","updated":"2023-12-13 01:23:44.000000000","submitted":"2023-12-13 01:22:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901582","meta_rev_id":"ac88e13b97705e2f1d8a4e2defed3e65cfca06cf","_number":901582,"virtual_id_number":901582,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-13 01:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-12 18:43:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-12-13 00:02:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-21 16:04:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 03:30:28.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"425f75c24623ed5a835a1a026b0ef06ff1ba0493","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-21 14:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea17b993c6f41dfe19bdfba38695c61c0a8e3d1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 16:04:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e113a58df36245859b585264c6121903\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/282d2f78eddb495eb1cbc42e7c2ebefe : SUCCESS in 47m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e2eb2ed226c94d02a2eec88bde77e057 : SUCCESS in 1h 09m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/95aeb89696164fa7b885a7eefb51cc7c : SUCCESS in 56m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3b9189dbf8774cd69aa5350c66a65d08 : SUCCESS in 54m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9340d3e9119846f7a120bfab754f5d6d : SUCCESS in 48m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/278dceb394944d59a2ad11f2f820b9c8 : SUCCESS in 1h 10m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3cd76994848b4996a429be243183bc5f : SUCCESS in 1h 12m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/71583b8296004df0b606463fbeb7c716 : SUCCESS in 50m 06s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f4d9b13c382943cc9d7ff2099efd52fc : SUCCESS in 53m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/687b76e4342b4a5ba13c3f039fcbf977 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe4d2b69f6b401da3613f28055591f7 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/717d2bca5047426fb6c49f3e679be036 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/430f10268c03452c8af2a7dab9e3e817 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae122af5177c4dd8b73ca17af6bb1c65 : SUCCESS in 8m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9cde6a87f77d481c9af4269a48ce8b52 : SUCCESS in 33m 58s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5971b147a0774f64ad311211450997e6 : SUCCESS in 7m 15s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7ee1f7239c4a417b889d1311744eb981 : SUCCESS in 34m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f98296f497f645b9b4c35b08fd191c34 : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e13369b1cb6ab27be44a31909337d50a9685ca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 23:25:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"47f1e6039215ecbc864e6f349e3e7a576ee894cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:34:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cba5f0140e24774a19375ff11616b21\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f902549f00414ae49eaa35c782cf7100 : SUCCESS in 46m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9cbe7d31c51c4ef4b5f80f886e3b3913 : SUCCESS in 48m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e64c0f5f0f534103b4718317ce0c8b6e : SUCCESS in 59m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/80efd0da18bf41aa9cd98be9795091e1 : SUCCESS in 30m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/85102d57ad7949ee88da6fdfb52dd748 : SUCCESS in 40m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a4420e90d4ae406683425688f1f2a7ba : SUCCESS in 51m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2dd113cb8fea47949b5706c89240ed60 : SUCCESS in 50m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b114163978944c8e9c68ac750cde38a0 : SUCCESS in 30m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/99b766b9ba734001be349e241c9275a5 : SUCCESS in 35m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da795660417447a294f7c51841b384c3 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9baa8dd8a4c485b9c4dd4615be03504 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd9b0e0547e84260a3e71181a7e3febf : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/336f6045406249c691e1fb99a78e22fc : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb918df3d7d64c73b1c46b28ddc6f3c2 : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0196b9ffd58942d3a7de8dbd7c35a9c8 : SUCCESS in 35m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/de44c51e2f3b4937a0f96d65b4af6d9e : SUCCESS in 6m 43s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cc5fedfd0dcb48aeabde0fd2ca22adda : SUCCESS in 47m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b24de9275b1a449e881a3753bb863ce0 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"afdb242c6855c485e7aaf0278c1ea1f499fe1096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-28 14:37:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d96080cc36abf9b600c61d3d60185e939a8c456a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 15:49:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a493bdcf8774033bb9f8937f5fca7fe\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9174d8cc9ee84378a03ef9c5dcbd0901 : SUCCESS in 50m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cc156724e72d4bea9926e0169c3066b1 : SUCCESS in 50m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6be6c144a744cd58aa23139fd688b57 : SUCCESS in 1h 06m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9bef78188a784d6792c932e26eecfac2 : SUCCESS in 41m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/420d03171dbc47178fa073044ee8f4c2 : SUCCESS in 54m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a60afa73f567472d8e8a8f0982b375dd : SUCCESS in 1h 03m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bbb0ac360a5f4ded85fa66e6584c5032 : SUCCESS in 1h 05m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d5fafc33e2884d40817b9138ef72147f : SUCCESS in 40m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c2e7a646b448457c800deb2dba115668 : SUCCESS in 51m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3eea315c1a244e9f9cffe4218daa47dc : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af31fbf2ffa747e8b9e380be353cb8a1 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7082cd3078194888b95e2e31bbedb9a8 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9a408ef640241838863660bbe518e90 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff0ff8bf1104eda9a0bd4363c57ac80 : SUCCESS in 8m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/260e635fb5eb42369c0cb752e0fdfe59 : SUCCESS in 34m 13s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5f01e3f84bb64d93b255404275d31ff4 : SUCCESS in 8m 25s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f579e378f11d4dcaaf2217845e7fd8e8 : SUCCESS in 54m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8595b52cc6354f8b98184d72596926cd : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"254abb2a2221d2a0eb81c79172c0f1878191c494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 12:09:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"618e68d94711380b75c3e3008513f14d21387359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 13:14:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c1ddbd9a583491eb9da0bd4dfff5256\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/86cc1dd190aa4c3f80d63d073c01b950 : SUCCESS in 35m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3b813ebf83054910b4098e745aefba1f : SUCCESS in 52m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/43c756398a4a475d8372f6df99abc214 : SUCCESS in 57m 50s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4d792210759c4225a5841af1b5a3f6ca : SUCCESS in 46m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f571c60588704fa08de5e5d04c1028f0 : SUCCESS in 53m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f60713ccf4f342fc91241387350fb6d4 : SUCCESS in 1h 03m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7481ad422ac4487b8accd1e7d2d77472 : SUCCESS in 53m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bcc642e9e1c14f26bfd6f297304fcea5 : SUCCESS in 35m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/058a4a8f1c464a27985e15da99b996c4 : SUCCESS in 41m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd2a82db8f1d4593a7cc8691415c1222 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f50df5bc0346038535108be34ee5f0 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e40a4d8f34b241abbc39ca55c3a31f10 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22903787226c4b27abec39061f4f889b : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5248418cd5436aaca4688e8d699bc2 : SUCCESS in 8m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b021cb5440646f69e267cc2b5a48234 : SUCCESS in 51m 34s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/157b65c3993e41f5adad238c8506bba6 : SUCCESS in 7m 20s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e51d6ff91931443aa45c0f33f41cc73b : SUCCESS in 37m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/29148a3bb04c481e99bf953bb4439188 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d1bf4d3b9494c357aeb72d1ad09b414f4da6dd62","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 03:30:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1e38ebdca69b156d5bd1868fac1f633566970141","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 14:07:01.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6d3b21c86164c2361e318f915f54a08ca9b42819","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 16:03:17.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f3eafe2f47c72501575e3bbcabbf8deb9fc20ba3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 16:03:28.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"820ca820c47cbffc011e3f9091df81d70dcd99c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 17:45:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8571ab19d69e440eadb27e13f056208a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/23802a74fd41459696def3a18ca4dbf0 : SUCCESS in 39m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/856ffd2211ea437fa8cb4d85cc6f98a4 : SUCCESS in 48m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aab484e1636b428bb7c66bb50fbf614e : SUCCESS in 1h 03m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4f4715305f4e4f3ba8b34cfdfabdc17b : SUCCESS in 45m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eaaadc100e9e4c8cad5e7bac7accb8af : SUCCESS in 49m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8a71eafbc8544f7eb5cc4716dda18311 : SUCCESS in 1h 08m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a1560372ad2e4f238aea2cce7417cf7b : SUCCESS in 1h 08m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9a2e189012e545949895a3765355afde : SUCCESS in 47m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3ae8bcc5045f400ba77ee5d12f0d3d0a : SUCCESS in 58m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad2b7721f9834c30a1c775ae5066c1bc : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133365c2932f4804bc2a86bb7ef18373 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa690d6b542d438d9fc6536b807d53b8 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fdb2513154e498f91bb1273b1d64691 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34cd1dc7993a487bb293a0994c32e16e : SUCCESS in 11m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ec50f2148324fc5a1dff7c89065666b : SUCCESS in 31m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1964ff26709e4cf39b29f27825a9e5e8 : SUCCESS in 6m 41s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/74af8da6cfd046abacf6b0d5427ac507 : SUCCESS in 35m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/09696a11355e4c40a3740cffb5cec9f7 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ba621be7c3c3a9ea19004567afc15dde06db1c51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 23:27:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"077bffc4765b61fe0ed80f84ef19c2d74b3fa2be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-08 00:43:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/531a30ca0a6d43b0aa437a084796381b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1651a4df0a7b457db0b2df9f90909d68 : SUCCESS in 50m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0d5653dd25c74c7f9ba79e3f29809fdf : SUCCESS in 1h 03m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d71158efb274bb8a02072c73ad10a16 : SUCCESS in 57m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8f5c3837d8524e039a23d287422d51ff : SUCCESS in 33m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/181bdb722062421cab05af44ebecb348 : SUCCESS in 38m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e31bf926629e45fe9dfeea63137d2f81 : SUCCESS in 1h 07m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b2abda5a2ef54a079d614c529f9ca998 : SUCCESS in 1h 09m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be6d1bc3d3164aaa8fe1e668b7699436 : SUCCESS in 51m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ba22a83c75f1406c8a2d8c1af14944cd : SUCCESS in 38m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bde2956307e45a0b724bb69bb77d794 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23947d129efd408983593febe0783e32 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2101019483824e0b823deca0cf6a6154 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d905037912604581ba61b5bbdd1705ff : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1fca67e8984d3489013ecc4653166f : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d766be1989f24d7e812ae0f8600dcd7b : SUCCESS in 32m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ab4e7985a48546cb99d101a6f6fdf0ae : SUCCESS in 7m 12s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/be0e31d2dca44568a4db0d73eb320f99 : SUCCESS in 48m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a6110967017e449c9a91dd6971680ace : SUCCESS in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c49c8c91d25dc51c29ab7d54ecbe738195e26c5c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-09 01:53:27.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"436f5ccfaa296826c16be03c264a22e0cf51238f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-09 14:53:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3a264cf6c1422f37388fbb2bb10c0d849b182886","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-09 14:54:33.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0fe2bbe57f78a2e0bfaa72cbd2315649384cdff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-09 15:02:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"2bd7bcefcca0e940427102667fe880fed68465fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-09 15:08:42.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"1a13a40144c0756e744334275f6b04af08784ea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-09 16:15:33.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58c3e745ac5c46fbb06cd460d31416f0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/70bdb0d1b12f4906ae10f182703c4bc4 : SUCCESS in 51m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/42473c2707934e098b914b0292b83954 : SUCCESS in 59m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e986c778bb984c7cb1dbdaa517093886 : SUCCESS in 37m 31s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a8129d41df964e1c912cb7a8f2286458 : SUCCESS in 30m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d82c1c1e7ebe409aa83a523eaa117d52 : SUCCESS in 34m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4a48620896534ca9a5bf636b0429bc5a : SUCCESS in 1h 01m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/347933cce9be4a3cbe66dbecceaa77fd : SUCCESS in 57m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d7a5e2a20e024126ae84aff17ab0847c : SUCCESS in 39m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e72798a5fa684dcc945d96db97a970c1 : SUCCESS in 39m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acee1199d340431aa55ec6915aeea4c3 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc4e1c97c444e7380adad0aeab2a72c : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0dc1b81ff844b1c8010da2ecb11b4e9 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f576fb28f4d348fe8b44c2b4ed494ee4 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1403e763441947aabb372a6e5298363e : SUCCESS in 7m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0696b22888604be395f8876f3b3f062a : SUCCESS in 34m 59s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/11dad58585d944eb83a53aa8cbb83c27 : SUCCESS in 6m 50s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/66ba562ce4df46aea539f4103ac0e407 : SUCCESS in 47m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cf02614e87fb4ff884c49bda52447110 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"cac1bcd05b202bc1289441e5496fae34c9ff7f84","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-12 03:59:47.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"36cbf250bd9ba24d771c1eba34fef4eb1037584b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 10:14:51.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3371e0073798770393465cd86554432232673476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 11:22:43.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f4221488e3048f89e836af60bac7386\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1910eb2e227c43f69bc879464c732d13 : SUCCESS in 51m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/17635ccdfad84b9caba9af6d70fc0b15 : SUCCESS in 45m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd4c23fadeec49dfa1ade3eb349249ca : SUCCESS in 1h 00m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/66653924682042b8963ffc2b43cb98a5 : SUCCESS in 47m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd6d76004c1a46c0b8b6312e0d26ff8f : SUCCESS in 42m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3c3bbd714a74461f91a2c498676aac81 : SUCCESS in 54m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ba9e516b5ff45cfb902f493b2088b1a : SUCCESS in 1h 01m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ab42d8a841da4ed996a49e7e64692ff3 : SUCCESS in 35m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6fb63e8134ba42ea9e54f76c92afcfce : SUCCESS in 53m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b808cb4a58445d4b1e77fcce6992b8b : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b679390349d04b998547d2d4c0576d6c : FAILURE in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ed8df9ed13749d591c11c4d4c99d521 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49444384d43f4a8d938ecc83c27bb06c : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7efcfd38f58540c996ac119a518cec43 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/aa8889a41cfe435683b17841fb529ff9 : SUCCESS in 33m 36s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/07346a5f9eee41549e80f9a87cde6483 : SUCCESS in 8m 26s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e7e3aed725ab42f1bbc08dfa2a6ad576 : SUCCESS in 36m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c7f21df82a9542b38b0620288f0acfae : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"3cdb1a40f92621511f38ac6aebedd37cb60e2788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-12 18:43:25.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"a629239d300b016d0c344576c9ddb39332244cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-12 19:51:01.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d5af212f8c4e0f958828a41203b064\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/974ec590cecc4362942774d159b398c3 : SUCCESS in 35m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/50f18ca8efd346f88041560de12adf9f : SUCCESS in 54m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b16a5916675742c68ee66491b71a1b6e : SUCCESS in 38m 34s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/44a032a73f054cd88c59d8c6e2b85d87 : SUCCESS in 48m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/46ae6d41bdae4b7fadc578f4a1e2033e : SUCCESS in 49m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2d82413b16f54285ac7779310080b208 : SUCCESS in 1h 02m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4559a8d058f7495db7206cca2e0d2670 : SUCCESS in 57m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e5e8ef6ab1b1477891ea9fe3a5456260 : SUCCESS in 47m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/26882dad27b5498297d7cbebeee72fa7 : SUCCESS in 48m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7f8ace897f64869bd36fe35097dc84f : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22fb16972f2944db8f2065353c00e368 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/743ca57cc91f4b43b6ef9cb730d435be : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/deacfdfc463f42219ffc7effe8d746f6 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2953ac58b544b9b6bf81ad2a6deed8 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b2a0ed7367f401d99a1b1e528eb9971 : SUCCESS in 32m 41s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bed8edc0123d469485b2367a3d6f9534 : SUCCESS in 6m 40s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dc159b1ff4994d69ad7efab21f3b4761 : SUCCESS in 48m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5814fe6c95d8447282c8fb6a241a5f49 : SUCCESS in 5m 17s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"22f08f8aea4a92052188f00ffdb5f5c63b1425c9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-13 00:02:21.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"37ba0768fcc035a3db0572d92650c6bc358b95a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 00:03:02.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"6e760ed85e13ff2a22da94bdecfed5505dbc596b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:22:42.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7579d35913064d3685b40ea4e2d6ecd8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a18725f0b7ce490bb99fd7cf67d6111a : SUCCESS in 47m 23s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/03a08ca228914db499d52557eaee5cec : SUCCESS in 35m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8379cc116fbb4f47a1ce66a55841ac77 : SUCCESS in 47m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/71c9226a6c324c98a684336959e82d2e : SUCCESS in 1h 12m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af9441f1e75848d29abecc762e208223 : SUCCESS in 36m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2ef41acb8efa47b58bca18d2565ee414 : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4211aaa32046445a8ffac20a929117e6 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c9cddc1bf2345abb83076882aba691e : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f530dcb7b5ca4829bb0ca6b669f3a080 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ee14c93f80494c86cf94ac7d4fd1a3 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cdd20b6f3e2e47d79b3e2b49b24bbd7b : SUCCESS in 32m 16s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f1e7690d99374a90bcbcd3d39bb556bc : SUCCESS in 6m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/78d4d314c1064d3f820d2c0e6943fe69 : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":11},{"id":"27f5b95744e4e0d941430aec533cdd60afc6b577","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:22:42.000000000","message":"Change has been successfully merged\n\n9 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":11},{"id":"ac88e13b97705e2f1d8a4e2defed3e65cfca06cf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 01:23:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5055c90b4d541fdbb9aab569c6819bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7579daf3ff148eea2bda652e4fa4c23 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d90691f3a29416dbb209b52aae288a5916bca632","revisions":{"0df8c9e9df7d807a6e5055ab0db1694db988c0b0":{"kind":"REWORK","_number":1,"created":"2023-11-21 14:31:57.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/1"}}},"commit":{"parents":[{"commit":"d0c8b250aa3bdde3d426418e3d3d359068c4d22d","subject":"Merge \"Remove unused base functions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d0c8b250aa3bdde3d426418e3d3d359068c4d22d"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0df8c9e9df7d807a6e5055ab0db1694db988c0b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0df8c9e9df7d807a6e5055ab0db1694db988c0b0"}]},"branch":"refs/heads/master"},"fec1225f4c0f5c68fe6a019c48c5a10e6ee8ea36":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-27 23:25:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/2"}}},"commit":{"parents":[{"commit":"9860624c48419c03ce6597ba0f9ae6d4a6443949","subject":"Merge \"Improved DNSMiddlware test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9860624c48419c03ce6597ba0f9ae6d4a6443949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 23:25:58.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fec1225f4c0f5c68fe6a019c48c5a10e6ee8ea36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fec1225f4c0f5c68fe6a019c48c5a10e6ee8ea36"}]},"branch":"refs/heads/master","description":"Rebase"},"0758531f689b34a5d520637f6906a8261e28ef4f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-28 14:37:07.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/3"}}},"commit":{"parents":[{"commit":"f9a2494b2db9a1ee63c9d8f72b54417846c6096e","subject":"Merge \"Remove executable flag from recordset.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9a2494b2db9a1ee63c9d8f72b54417846c6096e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-28 14:37:07.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0758531f689b34a5d520637f6906a8261e28ef4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0758531f689b34a5d520637f6906a8261e28ef4f"}]},"branch":"refs/heads/master","description":"Rebase"},"0add40e5e0ce43a52e5e23317a650668836abb60":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-05 12:09:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/4"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 12:09:12.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0add40e5e0ce43a52e5e23317a650668836abb60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0add40e5e0ce43a52e5e23317a650668836abb60"}]},"branch":"refs/heads/master","description":"Rebase"},"2867cdd8fea9e5dc47d58d714bd54432369aa9c4":{"kind":"REWORK","_number":5,"created":"2023-12-07 16:03:17.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/5"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-07 16:03:06.000000000","tz":-480},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2867cdd8fea9e5dc47d58d714bd54432369aa9c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2867cdd8fea9e5dc47d58d714bd54432369aa9c4"}]},"branch":"refs/heads/master"},"a448530abe8ebccc94e9182a33137fdab22f95d7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-12-07 23:27:59.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/6"}}},"commit":{"parents":[{"commit":"55cc882706668482e6a3538964b9609addc74964","subject":"Merge \"Cleaned up rpc loggingutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55cc882706668482e6a3538964b9609addc74964"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-07 23:27:59.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a448530abe8ebccc94e9182a33137fdab22f95d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a448530abe8ebccc94e9182a33137fdab22f95d7"}]},"branch":"refs/heads/master","description":"Rebase"},"44150adffca89530877f2cabb3fc69c34f9a75f6":{"kind":"REWORK","_number":7,"created":"2023-12-09 14:53:49.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/7"}}},"commit":{"parents":[{"commit":"55cc882706668482e6a3538964b9609addc74964","subject":"Merge \"Cleaned up rpc loggingutils\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/55cc882706668482e6a3538964b9609addc74964"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 14:53:24.000000000","tz":-480},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44150adffca89530877f2cabb3fc69c34f9a75f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/44150adffca89530877f2cabb3fc69c34f9a75f6"}]},"branch":"refs/heads/master"},"eb6e7e99d9fc3102469cc7b53e7c446d2e53b1b4":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-12-09 15:02:12.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/8"}}},"commit":{"parents":[{"commit":"4fb4182270434201532d63713a5584b66078dabc","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fb4182270434201532d63713a5584b66078dabc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 15:02:12.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb6e7e99d9fc3102469cc7b53e7c446d2e53b1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eb6e7e99d9fc3102469cc7b53e7c446d2e53b1b4"}]},"branch":"refs/heads/master","description":"Rebase"},"c184426ee2f245179ef0b72a8113d11dcebd924b":{"kind":"REWORK","_number":9,"created":"2023-12-09 15:08:42.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/9"}}},"commit":{"parents":[{"commit":"4fb4182270434201532d63713a5584b66078dabc","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4fb4182270434201532d63713a5584b66078dabc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-09 15:08:34.000000000","tz":-480},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c184426ee2f245179ef0b72a8113d11dcebd924b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c184426ee2f245179ef0b72a8113d11dcebd924b"}]},"branch":"refs/heads/master"},"aa669c728b7a7c564b6eb82798b13aadaa26dbbf":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-12-12 10:14:51.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/10"}}},"commit":{"parents":[{"commit":"53a96324c553b9b4e956d1f455af3aaa057b9ab4","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/53a96324c553b9b4e956d1f455af3aaa057b9ab4"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 10:14:51.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa669c728b7a7c564b6eb82798b13aadaa26dbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa669c728b7a7c564b6eb82798b13aadaa26dbbf"}]},"branch":"refs/heads/master","description":"Rebase"},"d90691f3a29416dbb209b52aae288a5916bca632":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-12-12 18:43:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/82/901582/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/901582/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/901582/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/901582/11"}}},"commit":{"parents":[{"commit":"2ffc4902a102110e272f69362086cd6e574b3aa3","subject":"Cleanup Central tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ffc4902a102110e272f69362086cd6e574b3aa3"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-21 14:31:37.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-12 18:43:25.000000000","tz":0},"subject":"Improved service coverage","message":"Improved service coverage\n\nChange-Id: I16ee9a86f64af2d666adc8a5cad5267fc34910b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d90691f3a29416dbb209b52aae288a5916bca632"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
