)]}'
{"id":"openstack%2Fdesignate~901594","triplet_id":"openstack%2Fdesignate~master~I76865c31211c8ad30e620444e470b39dffaebbbf","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2023-11-28 07:34:21.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-28 10:01:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I76865c31211c8ad30e620444e470b39dffaebbbf","subject":"Remove executable flag from recordset.py","status":"MERGED","created":"2023-11-22 16:40:21.000000000","updated":"2023-11-28 10:02:15.000000000","submitted":"2023-11-28 10:01:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901594","meta_rev_id":"d5475659f41b64a5ec61bae167114f394fcd3aa5","_number":901594,"virtual_id_number":901594,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 10:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-28 07:34:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-27 23:28:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-22 18:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 20:17:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"87e3ce6e9c836492f97b78f607df6d20199c32a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-22 16:40:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df1e1fa2eb4ee11f4e0ea8e5fbf1ddc8059c838e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 18:10:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b320431f5074d57a4d178448171d5d4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e9d0b2ae23024052a8bf49651f3eeabe : SUCCESS in 51m 31s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2366040b1a5444b08ef6360936e93e25 : FAILURE in 28m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fac188edb494c75a3487a225ac2142a : SUCCESS in 42m 51s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/450656ef831b404fb9c4273e02f70268 : SUCCESS in 38m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b5d957a9bd8743ac8f570b171d8cda6e : SUCCESS in 50m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ea6317dd13e8405a927593078100632d : SUCCESS in 1h 23m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1082dc8ace584b52851b384103ff4623 : SUCCESS in 1h 09m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/70033e2a32b4441d9d700c609e3f77af : SUCCESS in 40m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4d4eab1dc9d943cfacc46610853f71c6 : SUCCESS in 41m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d16abff9ca7c4062929ac8243f68bf32 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2e194d345041119ead17b34fc5a3c2 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d5575323b3b47e4962664049dac6be0 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6717bd62e2e47c3a049dcb01ff108f2 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd76cf1432d04facab339ced3e56696e : SUCCESS in 9m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b409aa381d534174a3dec86f987c9f94 : SUCCESS in 39m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/3cf991fc78354f5bb378713035f51f00 : SUCCESS in 7m 03s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6385998efb39487eb3eb6d10dbdf8212 : SUCCESS in 51m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a106e1aa3fd64bd5a081a23eac814835 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3999f70805b7683d74af5c750956fbee62dbc057","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-27 20:17:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76e59863a4c65da57b169432dc33bd8a3ee608b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 20:18:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d5637c1a667ea1e17c85019d3397b112f2a5097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 21:18:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abcff75c71e44c0187b7bf6575cb74c5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a09f58e5a64b43f785c1c5dd80ef651b : SUCCESS in 44m 07s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5de2c5626fe741a8a3d7848716c5cd40 : SUCCESS in 49m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fafae515793e4be5bc50c901edb59a77 : SUCCESS in 46m 18s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/569c0159088d4e4ba4fc991ae76fd8cd : FAILURE in 1h 00m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bf271ae1b01d49679ef743941f8c3e74 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a391e9a6d3cf40dd8e76e1dec82fc062 : SUCCESS in 36m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb2bcb3da8b41acb8f90e3c623fa615 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd2422b630cb4f158af39530019d03ac : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fad9484df54a40a398a2e681e0f3cc24 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c13c37d9210f4d518758ceca9f7ca390 : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c4388f1bf7748d1b9554307a625861b : SUCCESS in 31m 47s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fae2b4182caf4b758fba2ce4df945d3e : SUCCESS in 6m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5c3038bba2564d34a12dd9733fb22271 : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ebc83fca45d4f8e8976ffdc290fee268dc06a063","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 21:58:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"722d7c2238b4636bd8c945d06fe514c1a1e2632c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:25:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a0c9787f5c413da591ea97cf4eeff7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7530fbb073c1480d8a9fb1d68a1eb0ac : SUCCESS in 32m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d7f74b405bb5409188d4c2489433f012 : SUCCESS in 1h 01m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dc93b7a3056404883cceb80e0853ddc : SUCCESS in 40m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/eab3ebcf01c24e4ebe650e1cddcc7225 : SUCCESS in 47m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3778c47442de4b98a6a73abc570919ba : SUCCESS in 39m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0f6c750eef8541f8b38844cc64f430a7 : SUCCESS in 1h 18m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0bbceba607b04bb28838e73c7b31fb22 : SUCCESS in 53m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b5b48d27f3ac4f47929432518b4ef633 : SUCCESS in 47m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7facb31e421d4135a1c951889825acfc : SUCCESS in 35m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a89571e31f824e2c9a241776cd0a0a4e : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76ce4f9e47a549bcacdc080ad1ca37fe : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d713770065654f75b046de61f31df829 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2e2d999240b4821a6f613123c90ab9e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b3d618877224824afaa222655cb0fc1 : SUCCESS in 6m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f246af97dd15463392651e0d70aae16c : FAILURE in 17m 29s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2518afb4b8ea4c1cb8c3d3b556f195ef : SUCCESS in 7m 09s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ed4598acba0d4ac08dfa16d247ac0caa : SUCCESS in 36m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/028cf8d8205e44b7a27966d8c664bde6 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58dd0123a6796da722844a8533fd486b16049641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 23:28:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1722cb9c5da8b003b5cdb96023748d52371a23bb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 23:28:06.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d6bb811102977c9d99cd3fede969bed5e2265b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 00:47:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/995c6769801241aba0fe7b01d39f2f0c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c196ef2b39d64757b1a0bc8f07548151 : SUCCESS in 30m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/22add573421f4b5c9a0d69ee55699e88 : SUCCESS in 40m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/737d83000398456ba6d0fa399dc60921 : SUCCESS in 50m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2d5f3a2175914f2d981428670b05a759 : SUCCESS in 47m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/775ba81bc9204558bfb0ea9187edea1f : SUCCESS in 37m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9df7a550f4c047f193f0876e8d03b3e8 : SUCCESS in 1h 14m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb7b0bf95aea4b309c54dc46703ce2d4 : SUCCESS in 1h 06m 58s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d3034b66f5cb4d248673c9f6d257a73b : SUCCESS in 48m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cfd9cff4e64a49d89f728468ebb3ba57 : SUCCESS in 32m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/697949694d30437fb59eb09e0c434e38 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8473306fb97441aa9f3e2c8029e6ff59 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a5dc57cb5248138df2f54e3bef838b : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/569fae4f8d4644718375e02c436adbaf : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e69a9858d2d843389549075b4cbf0887 : SUCCESS in 7m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a60ff3940670409f8c969d2c97183cf8 : FAILURE in 17m 57s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/7c10de7ba846455ba71f60115f626c0f : SUCCESS in 6m 49s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/827f29324ecf4c66b9142cf617d5a910 : SUCCESS in 47m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2f2fd8fd351547dc82b0e8c35b726f67 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cb05b190cd1b1bc277a920af82de642be78a1757","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-28 01:02:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91bd80dd68d5dc14779ebb824cef690d4b8abb1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 02:28:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4567132b04b54a53bcef4700f2f3cf29\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/641b347d5df64193b371b9cf024ba986 : SUCCESS in 34m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b18c3387744a40a296bf711c6d578b7c : SUCCESS in 39m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/94f8f4812aa64eb9a491bc6231c1faea : SUCCESS in 36m 14s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/31d71ada88454ca598abb13959630dea : SUCCESS in 34m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f8b64befc654de09a055534d55cb9ef : SUCCESS in 46m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5e8052b4ecb04449ab2f2862f356fe71 : SUCCESS in 1h 17m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f940677264b5403fbbb45b4cbd345b8a : SUCCESS in 1h 08m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f1f2cd9b24b4e29be30084fa25c14b9 : SUCCESS in 45m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5627560043494b2a9ba6d416b0c976fb : SUCCESS in 54m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be4080a04bc74cb2a292e38ed46c3da1 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a6bb551aa94427a827a37750b9678de : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbf20efcd9d44d3fa1cd51e934b9be46 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/80c579035cbe4b84af1473a50941e1b3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebff25b207674e9bbb7a280883687d67 : SUCCESS in 7m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de8558e9b9df4391ab0b25ad62b8e0e8 : SUCCESS in 37m 20s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6c95bfe51685460bb53b5ce28c15f25b : SUCCESS in 7m 02s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/9679323b09044f9a8b6b9b6dca776689 : SUCCESS in 47m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ebe31de7327a4074ba76c1adf10c3b85 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c124a93a704d513706d46d9037e5e020be58841c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 02:29:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"950e82a1547e8ed9c40994a937e3ca02ed8a125d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 03:12:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0be0f1570d44fc484b9c3c66d24bef6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aa86f900d9074a18b8c423af52bf69a8 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3b70f0f3e8b446d7ae4ac846595dd19e : SUCCESS in 35m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cfce3399070c4750af574b2a79bef397 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68c75d1849b047e28ff7ce9e721994d9 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abfe803330d14f328f9d232ffde4e919 : FAILURE in 37m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5c7a056694ef4cf08a6541d5cbd1a066 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8066dc6d34e4484aa0df1848f321bf2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0de2f6275a984f10b70b9b6854199884 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fa930032204410bbebaf6dfc69fbb36 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f42f4af049c041eb981d4d86115beb11 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0d55aa8e9c4b4888a5338967f8b63709 : SUCCESS in 32m 39s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/12370a48964e4958be3ad873a866b566 : SUCCESS in 6m 53s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/471f73641ae04243bdbd7d6e24f1ac47 : CANCELED","accounts_in_message":[],"_revision_number":2},{"id":"cdefa590bd37f1a8d9c3a184c2c59e142f3669d2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-28 07:34:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44ba8d549cd3dde657c72fd9e98bcc70fd908371","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 08:49:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f55c9e2296a6459a90db82f1fc875361\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4ddafc77653244df92ad8d357fa5df1a : SUCCESS in 46m 34s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8dd8c1bca7a345b3a231183a639c29f5 : SUCCESS in 41m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e771ff47978436aa9c60552c15dcb82 : SUCCESS in 57m 36s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f9163546fb3f43219bbbac8d18d05e2b : SUCCESS in 50m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/47200f8cc47e4a60ab1c9d3ba3e7d812 : SUCCESS in 36m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/991d43957cbb4d878db76317b0e87de8 : SUCCESS in 1h 08m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/45db8e79293943daa385589a24e29514 : SUCCESS in 1h 02m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a4576ff8a33a4d3690c615f426f98820 : SUCCESS in 43m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/709d7e1e98ef410bbe08d50526dd97df : SUCCESS in 37m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da447c24d07f49e9ba37e39364c0206d : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d124bbba9b4557b60fca27bdd951ac : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2297e089e3764ddf9cdfc5e20b3c3627 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e46ae6c80bc74c3c88f8de6651ac5328 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395a7c2455e8441fb1609f926f0be980 : SUCCESS in 7m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/36ff0add1ccb4c1b86f82a82100da13e : SUCCESS in 43m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/cd7ea6cb215b44adb4a29f2988438c13 : SUCCESS in 7m 04s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/fcf4a9e04b064f8687e8dafe6d07ceb0 : SUCCESS in 40m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9c32d22ed2da40ad8385d10f9c29ef99 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4cd9897a354bce7deb91779658a8b6ca6dd9c336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 08:49:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"efb05fd9e5f122f72b297e4d5381641c88b9534f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:01:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a08ca84eb5054d169a332f5b9cffa2c2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/95085e50d52a4edaa87cf3ebdffd2400 : SUCCESS in 47m 27s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3262a24b31b94e92926607ec5d366c38 : SUCCESS in 50m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c35727492e542c1b3589e4d30c6f656 : SUCCESS in 49m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/88792774107446f591a05039ac413334 : SUCCESS in 1h 06m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dccea27ab83b4cad8c7a1c2e4b889521 : SUCCESS in 51m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3bd2c385f1cb4f4f94b06c4f56403b2e : SUCCESS in 53m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3233c3ab8054a5bb505aabdb0e37a2f : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbd7213aa1b041169e9aeee5c42a169d : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc264010d4c74f3a8a03591c7b3201a1 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/705dd80f5bc344489adb4c4ed15a80eb : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/964a7ccc7321491891536d7d7b4b1fbb : SUCCESS in 46m 49s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1772645821d8485b998d74f7b705d269 : SUCCESS in 10m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/55f3dca61b5b4787b5a8f514fc7f9931 : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4ac0b30a99e460f62ea04da73a8d4839ed9ce86e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:01:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d5475659f41b64a5ec61bae167114f394fcd3aa5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 10:02:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4704f3f1e55d4f85859a564fae3a74c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247895269fe64c269cd7c616b0894294 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f94164f224c62f526fb6c0d969b50a238c621f00","revisions":{"1b80300e6d5da4e4eed923738e91078b03ae4796":{"kind":"REWORK","_number":1,"created":"2023-11-22 16:40:21.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/901594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/901594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/901594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/901594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/901594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/901594/1"}}},"commit":{"parents":[{"commit":"21b266b070ab9d84e87aef8487dbc149e518c0cf","subject":"Merge \"Remove devstack support for suse\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/21b266b070ab9d84e87aef8487dbc149e518c0cf"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:40:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:40:15.000000000","tz":-480},"subject":"Remove executable flag from recordset.py","message":"Remove executable flag from recordset.py\n\nChange-Id: I76865c31211c8ad30e620444e470b39dffaebbbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b80300e6d5da4e4eed923738e91078b03ae4796"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b80300e6d5da4e4eed923738e91078b03ae4796"}]},"branch":"refs/heads/master"},"f94164f224c62f526fb6c0d969b50a238c621f00":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-27 23:28:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/94/901594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/901594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/901594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/901594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/901594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/901594/2"}}},"commit":{"parents":[{"commit":"9860624c48419c03ce6597ba0f9ae6d4a6443949","subject":"Merge \"Improved DNSMiddlware test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9860624c48419c03ce6597ba0f9ae6d4a6443949"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:40:15.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 23:28:01.000000000","tz":0},"subject":"Remove executable flag from recordset.py","message":"Remove executable flag from recordset.py\n\nChange-Id: I76865c31211c8ad30e620444e470b39dffaebbbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f94164f224c62f526fb6c0d969b50a238c621f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f94164f224c62f526fb6c0d969b50a238c621f00"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
