)]}'
{"id":"openstack%2Fdesignate~901938","triplet_id":"openstack%2Fdesignate~stable%2F2023.2~I71b0b1b237b5d5f12209f431db19cda1b44a1112","project":"openstack/designate","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-28 18:53:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71b0b1b237b5d5f12209f431db19cda1b44a1112","subject":"Fix Producer shard range ignoring the last shard","status":"MERGED","created":"2023-11-27 22:29:53.000000000","updated":"2023-11-28 18:54:27.000000000","submitted":"2023-11-28 18:53:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901938","cherry_pick_of_change":901592,"cherry_pick_of_patch_set":4,"meta_rev_id":"e004ec7a99a741d6105325180ba1a9d2f3b7ff6c","_number":901938,"virtual_id_number":901938,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 18:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-28 17:20:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-11-28 16:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-28 17:20:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-27 22:38:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 23:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 16:16:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-28 17:20:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"74e9965ea93c212b6042c9b99955a2681b045e5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 22:29:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"19f915bca7c8e95362f3071816a0729979ba72ed","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 22:38:53.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"68a1e00bbccc1d417c2a569af2032e90dc7ed8cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:44:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b72275af57634d7fb461b1bbfe7eacba\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/280d31876ecc47d39d5d0fd07025fbd1 : SUCCESS in 36m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9b650b16816544cd89597b6895c3c783 : SUCCESS in 42m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/51500cfa290847fabe92ef0074d00368 : SUCCESS in 41m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7b38869f68104ec5a9040f51a88771e7 : SUCCESS in 40m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/38ad1eaea01640eb9d334b1f141fc33c : SUCCESS in 41m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b3acb6664eb046f68c8ea6e96fca0cd9 : SUCCESS in 1h 04m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/08d2b17217a84996b6a8f064cf61d643 : FAILURE in 1h 12m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/06baebe543b44d6d938dcd7a84bdfb91 : SUCCESS in 42m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6c1cc667337d40df89b9cca709a2a1e1 : SUCCESS in 53m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2778d032bc1c45e2a1b355725c873b8e : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1def72b39bcf472db132cadfdd8b44d7 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cd3b9df8b1947a7941ce4750a2b49f2 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/099d7cca2f704b838e8459aed7f99fc6 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ff1da2b6b224a2cb7709348581ac235 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee5568c1bb824ff895f58a9ed109b963 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522a2cac3ef54baf849a48bed5c5f70f : SUCCESS in 7m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e0dd2ad9f519493da955fa10523e0503 : SUCCESS in 40m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/544781df9a72494ebcdcd66155320029 : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b11778bf4b979994e829462c49b61ec156845c13","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-28 01:03:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a86f34f62fbf038c1839da93ca3cccbbb7ada5b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 02:20:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e9c7c0e2c94cf182a3e6d72d0b8716\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/af14cd5f2ddf40ed9a47381963ea3822 : SUCCESS in 36m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0485e180cf3f4c6a9b64a66c975a90b0 : SUCCESS in 49m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e774cee24a07430cb6472d014b5dd881 : SUCCESS in 41m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4c10cd8d58504b199679548e89a9b40e : SUCCESS in 35m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a5e36419aa7d40afaaec9c85c6ca93f8 : SUCCESS in 41m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dc63b6f5cacc4fc18d38bd57cd0cdce7 : SUCCESS in 57m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1783b913d24b4b5db724d8a6988aeaf6 : SUCCESS in 1h 11m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0c6784c1c56e4084b8bd9b962ded58e9 : SUCCESS in 52m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fa17eeca042641c9a40c1e975bb16dc5 : SUCCESS in 49m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/407e4c44d771419a9d6424aad4fe64ce : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4671404674644b6bbc659388417cfd0 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f425539292e4bba9bef3e35ea04233a : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77d77c58c6244dc98b212e1d2f202c86 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13193ec23bb04370987b8c56f8f91dd4 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7126f6c1af844c94aa802f0f647533c2 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3a63af8ab3d467997c12fe4282aed71 : SUCCESS in 9m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9ddc4a2372f540c89b9d4f99b073a238 : SUCCESS in 32m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a15fda45e824434594459ab69fd1dcf1 : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c40abdec3c775b091808207d022beafb39f2183b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-28 16:16:56.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70f46169f2f1a2258f4ff499e3867d3fe638554d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-28 17:20:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93c72c980a607bf497af8c9edacdb50f4b53cdfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 17:21:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4527605360cf47cc67129290b9c38f2116c365ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:53:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97264207d51142b0a477d43080cdf846\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/085d2f8048954e37b37197004de6257b : SUCCESS in 52m 27s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/b44174dc93a44a98b887a56cf256c1d9 : SUCCESS in 38m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/83c77582bc2c4413a7e184974b145997 : SUCCESS in 45m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c254e2389624ddfb0ce24e2f49f8911 : SUCCESS in 1h 13m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/58953534754c4e24b37aa77b443fd8f7 : SUCCESS in 36m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d3603ec0f3384f288dfb3d32a4615484 : SUCCESS in 54m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5516d1d95f4ecbb7a8fa1fcad3ed13 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87758f8d01d545db8fa83753917e5c33 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a010574fff445638aecfeeccd55821a : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5df36a258124ec69097ce09207e1cb5 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c4aae8b594040a29e52f200c8d5472c : SUCCESS in 6m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/77089b33122543c6bf2372d484ed6540 : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2cd8134961e44b3a75f70bc7de5e51ac9808ef18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:53:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e004ec7a99a741d6105325180ba1a9d2f3b7ff6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d72bb71aee3042688c7d8f1eae16253a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b4f2d154ad42e8b3540ac919df3b42 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78a53cde57e17fde59d5868255878ef19c64bcac","revisions":{"78a53cde57e17fde59d5868255878ef19c64bcac":{"kind":"REWORK","_number":1,"created":"2023-11-27 22:29:53.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/38/901938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/38/901938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/901938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/901938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/38/901938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/38/901938/1"}}},"commit":{"parents":[{"commit":"3de4f7f24b02fb77d20846daf7a7c9b9c56e1dd0","subject":"Fix bug with new zone default values being ignored","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3de4f7f24b02fb77d20846daf7a7c9b9c56e1dd0"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:10:49.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 22:29:53.000000000","tz":0},"subject":"Fix Producer shard range ignoring the last shard","message":"Fix Producer shard range ignoring the last shard\n\nWe define the shard range like this in objects\n\u003e \u0027shard\u0027: fields.IntegerFields(nullable\u003dTrue, minimum\u003d0, maximum\u003d4095),\n\nThe problem is that in code we handle it using range(0, 4095), but that\nrange does not include the final shard value of 4095.\n\nCloses-bug: #2044278\nChange-Id: I71b0b1b237b5d5f12209f431db19cda1b44a1112\n(cherry picked from commit 54d2d2cf187aec8d73bd7588dabdc4279ca8851b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/78a53cde57e17fde59d5868255878ef19c64bcac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/78a53cde57e17fde59d5868255878ef19c64bcac"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
