)]}'
{"id":"openstack%2Fdesignate~901939","triplet_id":"openstack%2Fdesignate~stable%2F2023.1~I71b0b1b237b5d5f12209f431db19cda1b44a1112","project":"openstack/designate","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-28 18:53:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71b0b1b237b5d5f12209f431db19cda1b44a1112","subject":"Fix Producer shard range ignoring the last shard","status":"MERGED","created":"2023-11-27 22:30:10.000000000","updated":"2023-11-28 18:54:27.000000000","submitted":"2023-11-28 18:53:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901939","cherry_pick_of_change":901592,"cherry_pick_of_patch_set":4,"meta_rev_id":"5ca0b05ad4ea61082443381b525df6e95687b56b","_number":901939,"virtual_id_number":901939,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 18:53:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-28 17:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-11-28 16:14:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-28 17:25:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 23:42:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 16:14:23.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-28 17:25:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"cfee1077ada0b503b0938e11002bdf2f5ff93e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 22:30:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f753ed24e68ef48571289e8b6a36cfc116ff03f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:42:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6680a43451c0470f8da93df93fc8c555\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a7522ccbe52d4edb997387276da4095b : SUCCESS in 44m 58s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ab692a028d344419adb8e268525b14cc : SUCCESS in 1h 03m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/aa398704f7c848a08cd8dbbb2ba92ae5 : SUCCESS in 53m 11s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/4b2a7ce4409a47c8a2b060a7c9875129 : FAILURE in 46m 57s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ec14745b83054d4eb87e2ab508c7d100 : SUCCESS in 27m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9fc1fd25ed28443e939f84aa50ad1b3c : SUCCESS in 58m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/93858cb87eb149d28f410a19ba335b6a : SUCCESS in 54m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea23fb537e01443a9ea014f11ff5618f : SUCCESS in 45m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5eaec331ce374776a2c28b719e1b1189 : SUCCESS in 45m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccac1c0543f44222aca681237d45bfa2 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8375e9dcf0f34264b40e38487fc73622 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55c65914a0074eafbe736b186f448bec : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa91bba8d70c46be8dfe9eb6c8274181 : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/16488ce88b3d473bb5b5aecc1eff1536 : SUCCESS in 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6efa60096468bd07ebf279d497f2973cefa8c8f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-28 16:14:23.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1938dfad0ac13b9ce593bd8443822fa5af95976c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-28 17:25:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14f2b879c4bfcb375990c101b471ccfb88bd152b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 17:25:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffa87458c903c8ab8cd9bedbd50aab03413707ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:53:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e849052af4664c3588517d14c0cd965d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/55b31683ef87403987d0526f0635a45d : SUCCESS in 44m 50s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/93624ffdcceb43ccb1154850a951b342 : FAILURE in 30m 53s (non-voting)\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8cf2db15139941c299827f41abe83a28 : SUCCESS in 29m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/224b7f458e734f8a8f8a638e42ae07b4 : SUCCESS in 49m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8584658d987a452a840535d2f5b43e0f : SUCCESS in 30m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7a43b47577924ffaa3d22388a87ccf46 : SUCCESS in 33m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9c6e8c2ff0462cb811a3bf02890a83 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6479b60f03364f778be999708283e9e2 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db0af06663424c03bd6ceb692b724e6d : SUCCESS in 8m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d521867f5db940809cb8511116356899 : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"daaaa216df657144ba922ff18b20c22367078774","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:53:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5ca0b05ad4ea61082443381b525df6e95687b56b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de2e1ffa4dd14c6d96b32f5beffe59a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/419b3c9a61ef4b91b2d35b3dda0e4042 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff1986cbb755f0d74a7db98623d6d17816d933a6","revisions":{"ff1986cbb755f0d74a7db98623d6d17816d933a6":{"kind":"REWORK","_number":1,"created":"2023-11-27 22:30:10.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/39/901939/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/39/901939/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/39/901939/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/39/901939/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/39/901939/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/39/901939/1"}}},"commit":{"parents":[{"commit":"762bb3e8478469c9103aa6544e75ed9c5a5564f5","subject":"Merge \"Add middleware to limit request body size to api-paste\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/762bb3e8478469c9103aa6544e75ed9c5a5564f5"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:10:49.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 22:30:10.000000000","tz":0},"subject":"Fix Producer shard range ignoring the last shard","message":"Fix Producer shard range ignoring the last shard\n\nWe define the shard range like this in objects\n\u003e \u0027shard\u0027: fields.IntegerFields(nullable\u003dTrue, minimum\u003d0, maximum\u003d4095),\n\nThe problem is that in code we handle it using range(0, 4095), but that\nrange does not include the final shard value of 4095.\n\nCloses-bug: #2044278\nChange-Id: I71b0b1b237b5d5f12209f431db19cda1b44a1112\n(cherry picked from commit 54d2d2cf187aec8d73bd7588dabdc4279ca8851b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff1986cbb755f0d74a7db98623d6d17816d933a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff1986cbb755f0d74a7db98623d6d17816d933a6"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
