)]}'
{"id":"openstack%2Fdesignate~901940","triplet_id":"openstack%2Fdesignate~stable%2Fzed~I71b0b1b237b5d5f12209f431db19cda1b44a1112","project":"openstack/designate","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-11-28 18:53:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I71b0b1b237b5d5f12209f431db19cda1b44a1112","subject":"Fix Producer shard range ignoring the last shard","status":"MERGED","created":"2023-11-27 22:30:32.000000000","updated":"2023-11-28 18:54:27.000000000","submitted":"2023-11-28 18:53:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901940","cherry_pick_of_change":901592,"cherry_pick_of_patch_set":4,"meta_rev_id":"e6d932708f2307cac1aaac1d73eafce10fd9abd3","_number":901940,"virtual_id_number":901940,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 18:53:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2023-11-28 17:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2023-11-28 16:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2023-11-28 17:27:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 23:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 16:13:53.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-28 17:27:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"934b75c179631a3522cf283f2e75bd2fce992e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 22:30:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2150dee2374d0fd5d1e70a5b29147d675e24cd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:34:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb6d7fd52bc41f5b1412d1db298cc83\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e793d0abef584823a6e4e309e849ed87 : SUCCESS in 34m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8976262c68d44e0aa2f5421e40ec3c7d : SUCCESS in 33m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f171b549ec94d989d13eac72749273d : SUCCESS in 58m 54s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/80924a684f634b75b605a9ce3ed4912c : SUCCESS in 46m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f259ac95941245d99516c3482a049b2f : SUCCESS in 43m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/39cf17944edc4d158664ac16040b323d : SUCCESS in 1h 01m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/62c3ce6298e34abe9adaa033eac83705 : SUCCESS in 1h 02m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/52fd186bf2db41058a3f108e57b16ae7 : SUCCESS in 44m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8dccc53790174c8d99e2f0cbfe41b83c : SUCCESS in 47m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64dbaec0a85d447bbc0a7cc03d1895c7 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9163bb4e8f84fed9d33c3c0df3a26d5 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51deda93caab4aae8d5ed9fbb88855d0 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/430a7c52cf2c4f45855d4b008a886286 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21fe5d4b6cba43dd9602fe9cb9c926a9 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c862945ad314262857fa716292755e2 : SUCCESS in 8m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/57f511f3322a4de99fdf442cba079bc8 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8d1e764f9fb56bae786d5e76f1be78264afaab54","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-28 16:13:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da93c9c27a3dd32c1f5ad4fd7785c2b26cfa016a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-28 17:27:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b0aede4bb5f674ece6454d29e82e7c9a40505a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 17:28:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"03111154fd22256191f8fae30606ad90af4e67b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:53:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/990d4259a5d24a9aa5f632b2d32278cd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/540c5a599ccf469eb9b79b70da7ee67f : SUCCESS in 37m 14s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/63dd24ef14154301b7e8198851267cc0 : SUCCESS in 32m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd2d6026ec1a462aaf1f961ea3c7f6f1 : SUCCESS in 46m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/203b508955914412aa382319eab01492 : SUCCESS in 54m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/38c4b7089a954ed8a28c316f9c5ee640 : SUCCESS in 43m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fbbb0ddc2be840d48a77c4c7676b8085 : SUCCESS in 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/149aa3e17501404b8a4d467109b699a0 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5acfb627f7a49329eec9cbdb88e90de : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7deee349f6d34f54b448a563930c3476 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/924eab158266470d9c5dde020050e351 : SUCCESS in 7m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d61ac10e0f544d1bbcaa9130532a0976 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a757a56f83b2cab27cc6d6a210031a7c4a2c8fe4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:53:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e6d932708f2307cac1aaac1d73eafce10fd9abd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 18:54:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9eddf0c6862747e1a72973704e2b03fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e3ca1cf38d4ecb87d655bc219c773e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d247267823034c5e656f74e91b50475aa54d3fa6","revisions":{"d247267823034c5e656f74e91b50475aa54d3fa6":{"kind":"REWORK","_number":1,"created":"2023-11-27 22:30:32.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/40/901940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/40/901940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/901940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/901940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/40/901940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/40/901940/1"}}},"commit":{"parents":[{"commit":"4012425843529b02486c39421dc7593fe1803367","subject":"Merge \"Add middleware to limit request body size to api-paste\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4012425843529b02486c39421dc7593fe1803367"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:10:49.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 22:30:32.000000000","tz":0},"subject":"Fix Producer shard range ignoring the last shard","message":"Fix Producer shard range ignoring the last shard\n\nWe define the shard range like this in objects\n\u003e \u0027shard\u0027: fields.IntegerFields(nullable\u003dTrue, minimum\u003d0, maximum\u003d4095),\n\nThe problem is that in code we handle it using range(0, 4095), but that\nrange does not include the final shard value of 4095.\n\nCloses-bug: #2044278\nChange-Id: I71b0b1b237b5d5f12209f431db19cda1b44a1112\n(cherry picked from commit 54d2d2cf187aec8d73bd7588dabdc4279ca8851b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d247267823034c5e656f74e91b50475aa54d3fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d247267823034c5e656f74e91b50475aa54d3fa6"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
