)]}'
{"id":"openstack%2Fdesignate~902022","triplet_id":"openstack%2Fdesignate~stable%2Fxena~I71b0b1b237b5d5f12209f431db19cda1b44a1112","project":"openstack/designate","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-05 18:33:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I71b0b1b237b5d5f12209f431db19cda1b44a1112","subject":"Fix Producer shard range ignoring the last shard","status":"MERGED","created":"2023-11-27 22:31:06.000000000","updated":"2023-12-05 19:51:31.000000000","submitted":"2023-12-05 19:50:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902022","cherry_pick_of_change":901592,"cherry_pick_of_patch_set":4,"meta_rev_id":"9a689a2f6b2f619390d86fd8f8fb0b0e8285ee4b","_number":902022,"virtual_id_number":902022,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-05 19:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-28 16:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-05 18:33:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 23:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 16:13:38.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f7a59ecef1d45f54776b07e2174268f84810d9f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 22:31:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4e5d5167a82d352d232605c82b3f5d34784fcc02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:34:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ccc0539a4984c2b8544afa7c46249a2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84f3c4ccafd64e06ab40f06b0e55915e : SUCCESS in 53m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b1da165f5ef44e21ba41621ea600c7f8 : SUCCESS in 58m 21s (non-voting)\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/ac08f8de038c466585abe8fe4f451934 : SUCCESS in 53m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/439e04e8caf944028b9a1e3a30570720 : SUCCESS in 40m 13s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f858bc66e47440318a10f804648dfa29 : SUCCESS in 54m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fd3980f328ec413fba26d9cf17f3f864 : SUCCESS in 41m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e522ed97ef0e4605b340bc8707195af2 : SUCCESS in 48m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10c1b08776784d8eb297dfb8c00287cf : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dea876db3324d2ca33921b88a033fcf : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/672f464af646488c97d7312a8b8b982a : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dd35b0d261c48f88852c39ac6b0bf23 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7cf87ff91fa46a2b6c706bc407ac630 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310602d5227842ff832578c3c49e7550 : SUCCESS in 7m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/711d55f7766e409ab5b3e788c51cc11d : SUCCESS in 50m 12s","accounts_in_message":[],"_revision_number":1},{"id":"37496dc224f396a6d35bc444fbe37c5ba02f032b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-28 16:13:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a6b32dd96d11398ee2fe681af409eccc1795341","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 18:33:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e7eed4666a3e2c01fb1b687300b182e9ffce80d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 18:33:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3db27acb79220998f0b8f9ff61e710b54d26d349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:50:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3afee2b58e8e4f4ab427d2d39d30c5d4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/87bbc6bb6d1b4bc5902ea551b7c19403 : SUCCESS in 48m 27s\n- designate-bind9-scoped-tokens https://zuul.opendev.org/t/openstack/build/9cafd771cece4dcfa0320624c9e27ba5 : SUCCESS in 52m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/acc0c304158247de91939cf82b5c306b : SUCCESS in 53m 30s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f7d323268fa413884c503b99fd5b33c : SUCCESS in 1h 07m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8fe7db252f9e42b2abcd90d2c565dd44 : SUCCESS in 48m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f9b182a16dc344e8856813068af0d7d6 : SUCCESS in 52m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7b00eefc944f429caa0c8391acb8e3 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da848fcadd65468faa3a73c9c306aaf5 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a164009c51ae4069999071f7fd71a00d : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06367c5feba4bb5bf13825dd0176aa8 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e3cbc6943f75e10dfd04744f6a0bb8deab7f9f9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:50:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a689a2f6b2f619390d86fd8f8fb0b0e8285ee4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:51:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20a13b8b1ce47f5b5f813aafb27771f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84adee6db87b4790a31491a25d5e73fa : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64655c9076a5487fc2dea59a109c7d0d5201b0e9","revisions":{"64655c9076a5487fc2dea59a109c7d0d5201b0e9":{"kind":"REWORK","_number":1,"created":"2023-11-27 22:31:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/22/902022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/22/902022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/902022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/902022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/22/902022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/22/902022/1"}}},"commit":{"parents":[{"commit":"f5e59997f6d0aabe20686802c8d1a98d2bf7753a","subject":"Merge \"Add middleware to limit request body size to api-paste\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5e59997f6d0aabe20686802c8d1a98d2bf7753a"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:10:49.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 22:31:06.000000000","tz":0},"subject":"Fix Producer shard range ignoring the last shard","message":"Fix Producer shard range ignoring the last shard\n\nWe define the shard range like this in objects\n\u003e \u0027shard\u0027: fields.IntegerFields(nullable\u003dTrue, minimum\u003d0, maximum\u003d4095),\n\nThe problem is that in code we handle it using range(0, 4095), but that\nrange does not include the final shard value of 4095.\n\nCloses-bug: #2044278\nChange-Id: I71b0b1b237b5d5f12209f431db19cda1b44a1112\n(cherry picked from commit 54d2d2cf187aec8d73bd7588dabdc4279ca8851b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64655c9076a5487fc2dea59a109c7d0d5201b0e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64655c9076a5487fc2dea59a109c7d0d5201b0e9"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
