)]}'
{"id":"openstack%2Fdesignate~902024","triplet_id":"openstack%2Fdesignate~stable%2Fvictoria~I71b0b1b237b5d5f12209f431db19cda1b44a1112","project":"openstack/designate","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-05 18:33:01.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I71b0b1b237b5d5f12209f431db19cda1b44a1112","subject":"Fix Producer shard range ignoring the last shard","status":"MERGED","created":"2023-11-27 22:32:01.000000000","updated":"2023-12-05 19:49:49.000000000","submitted":"2023-12-05 19:48:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902024","cherry_pick_of_change":901592,"cherry_pick_of_patch_set":4,"meta_rev_id":"b41157d023df12c4b090da27a3258947321db143","_number":902024,"virtual_id_number":902024,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-05 19:48:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-28 16:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-05 18:33:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 23:38:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 16:14:09.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4b28324112960850ce51f45360bad54409142d42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-11-27 22:32:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9eb7074bbc5db699b447ff536b2abc138659c35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 23:38:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ed11597c7c4c0287d64cf5728f92d6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/228f9921c17449a8a5842dde7b56f0af : SUCCESS in 29m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6133573216f64f3f95e5b7ebfb57d46d : SUCCESS in 31m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d26f50e7ca745cb8952200af83a202e : SUCCESS in 1h 05m 42s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0073f48d38524cdb844bf5411b8d26a0 : SUCCESS in 45m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5c6a71cf6187491cb2e3472f26e4d38a : SUCCESS in 43m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d4fa2bca32d4e70861c5eefa2350e9e : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d31779c3d5d44d4a9b4cc6d3bde459d : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e26770fc10264116b4bdcf20b7967aed : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e763a6bd393d406baac1c6fa7c9d4c18 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01a0229208934dd38701e2f2995791e2 : SUCCESS in 7m 05s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/8570b97c716f4bba81e80468422ee927 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bd81de81022a91f5ffeec2a10f9df62be999d275","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-28 16:14:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eadaeae029977f2f0e51afd3c3457707e23464f7","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 18:33:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"844b7f7cdc5abfa5905de293a42e5b21c7ead849","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 18:33:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abd1733eaec5ea79363f2340f446df7b926346fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:48:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/034cee464a004b61a32c32899a32b934\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/963cf5549e2c4bd3bf5fbc0f5c4f5c43 : SUCCESS in 46m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/86f856a9c88346aa8d4c12471d849910 : SUCCESS in 43m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e7eeceaa93bc4d3096ceb7bef8fb650a : SUCCESS in 45m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0744a3bc5e5f448facedf92c9f3aebd0 : SUCCESS in 43m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6ba5a1cc3994e708b83dbcdfdfb9927 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1199aef4ac9245a78d64a0a6adae95ed : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c25a84ad0794ebca90ad70112796050 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9186e4379f16496aa98705cc0c8b1706 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":1},{"id":"ade5aff989ec17824b544d94eebfcd7b196c98dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:48:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b41157d023df12c4b090da27a3258947321db143","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 19:49:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abbf6256368640a9957113dceec23b6e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e642ac2839e94fb59c348857ee52266a : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1cd253895b8cefad3a2fa662543f88185151309","revisions":{"b1cd253895b8cefad3a2fa662543f88185151309":{"kind":"REWORK","_number":1,"created":"2023-11-27 22:32:01.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/24/902024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/902024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/902024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/902024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/902024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/902024/1"}}},"commit":{"parents":[{"commit":"0ffc851d2f97ea275402db313d0afd240e4003f1","subject":"Merge \"Fix designate-manage pool update bugs\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ffc851d2f97ea275402db313d0afd240e4003f1"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-22 16:10:49.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-11-27 22:32:01.000000000","tz":0},"subject":"Fix Producer shard range ignoring the last shard","message":"Fix Producer shard range ignoring the last shard\n\nWe define the shard range like this in objects\n\u003e \u0027shard\u0027: fields.IntegerFields(nullable\u003dTrue, minimum\u003d0, maximum\u003d4095),\n\nThe problem is that in code we handle it using range(0, 4095), but that\nrange does not include the final shard value of 4095.\n\nCloses-bug: #2044278\nChange-Id: I71b0b1b237b5d5f12209f431db19cda1b44a1112\n(cherry picked from commit 54d2d2cf187aec8d73bd7588dabdc4279ca8851b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1cd253895b8cefad3a2fa662543f88185151309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b1cd253895b8cefad3a2fa662543f88185151309"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
