)]}'
{"id":"openstack%2Fdesignate~902501","triplet_id":"openstack%2Fdesignate~master~Ie7a00355b853a17d06131d46b17003650126c109","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-12-07 22:34:29.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-07 20:06:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ie7a00355b853a17d06131d46b17003650126c109","subject":"Cleaned up rpc loggingutils","status":"MERGED","created":"2023-12-02 15:09:05.000000000","updated":"2023-12-07 22:35:30.000000000","submitted":"2023-12-07 22:34:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"902501","meta_rev_id":"96cf87c41f20714742726749dcd776df15c3f170","_number":902501,"virtual_id_number":902501,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 22:34:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-12-07 03:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-07 05:56:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2023-12-07 15:50:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 16:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 03:55:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-07 05:56:20.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"1947a0604f8f0eece1831d4afb541f6eb51c8d6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 15:09:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01de225d0b5d4a746455e1749b323bf02084f64c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 15:33:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc99b8e994e4c50de586b64aadf45c337abf9383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 15:38:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"35fc7f3cdbc17be2b35b778e8b097efdc34d5cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 15:42:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ffb6fbe17597261172bb366305b9ac0e2d1e14bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 16:49:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96ce1faea0b34633883ec12d508b4331\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8fc947b569114d15ad377766f59dd672 : SUCCESS in 33m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6aade705963b4ce4bc9c6fa91330ed68 : SUCCESS in 39m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f969dd9b235473ba4e9e5f74459c34c : SUCCESS in 42m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0f143540087e4a2e88da674920a8d1aa : SUCCESS in 37m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/03d9c6064c204233b0a61bc49cba988d : SUCCESS in 36m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d20cf4201d7d47dca2a11a2d0f39cf85 : SUCCESS in 59m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4d4153f455d4437db3b3c08ccf7007d9 : SUCCESS in 59m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3a38f70e775846c0ad5d4cdf0640c81b : SUCCESS in 35m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9be8141ca08b4c52a715b52f408754f4 : SUCCESS in 37m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddf0aaae066a485fbd1ab1a4ad7c26f4 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e5721c5fd464d0db7112df7cc5af41a : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d71f140d9799449da4ba0e5eadd9e7da : SUCCESS in 7m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6ae3c24335342c48b379d34dfc549cd : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ba86d7483714197a9b4fd8e429517f4 : SUCCESS in 7m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/083554fca06847acbfab6e4a15d1a5a0 : SUCCESS in 34m 06s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a2a8b010280a486b9233d68b1e2a3c19 : SUCCESS in 5m 20s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/1a5f9cb70d48477ca4f2e323938eb6d9 : SUCCESS in 50m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/727341df551f45f9a0817c9bb75bc1f0 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c5cc1e060480218d0742f76725953e918092a469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 11:39:54.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"61f9f6925ddeeb9057be680a4acd385903d79ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 12:47:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cfe3b7a58ba4bffb82cbff0c892a0f9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/88d3d4689e7846d2b45c44e47d30b566 : SUCCESS in 50m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ad428401ba9b41799bfebdd5d23df6cb : SUCCESS in 1h 01m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f9ef4c0657c447eab41e2c36a2fb4031 : SUCCESS in 57m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c66b27f8ee424701bc69d0e4510cc87f : SUCCESS in 28m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f6cc963019d42d09b8380187c4ee991 : SUCCESS in 34m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0365dcbe2ade40a2940999dbf8e1df9b : SUCCESS in 53m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/740ffa164f3c41ac9300c30644358016 : SUCCESS in 1h 00m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a053bd3aedcb4befbc3ac5e14fa334ed : SUCCESS in 40m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c9ff32f68d3b465dbde626ec6259d003 : SUCCESS in 40m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b34215834a04225963c0c817b372b5e : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9cd2d136f546db967924f6a5480a45 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/960c4a7e324a475ba008af6f110def5e : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/79a283cffcff4f4189a5d56848ee8a68 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d21331a695e4c309d4f05ff66d3209a : SUCCESS in 7m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/510f09bfdbfa412895db3d90189f946c : SUCCESS in 37m 37s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f3dd9eb017164448afa2cf948bd7a744 : SUCCESS in 6m 37s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/dd394e33b9a44bbca5ff37c898ff1550 : SUCCESS in 51m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/08c19f6a33364724b0c65fdb7ba6d281 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d19071824d9038f78b80072ed0e8cb1bab668726","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-06 13:03:54.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"8d1e4f2cadd0a894cfa55d7f1545881398b26b8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:25:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ca808079124cada793baef432e3f52\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ca08ee98787b46238481344c3a141821 : SUCCESS in 42m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4627a2d6cd35485d8459404cb029fa91 : SUCCESS in 46m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4d30f0c048a4c95b0b5d8d46c026d30 : SUCCESS in 45m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e9a77c0598644648b601487e976fa2a0 : SUCCESS in 46m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5beeef20fc544521adb4fdfd24fc5bb5 : SUCCESS in 48m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ee554893b4e1478f8cff98e0f674a72c : SUCCESS in 1h 09m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7c06ec441bc44b28f8ab8cd18fc5fbf : SUCCESS in 1h 12m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/36a142a4dfb74d7b9e09b435865e65a6 : SUCCESS in 34m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/67a44b2771d9490a9c8182b13ec4b5b9 : SUCCESS in 40m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f082f96468e84bb9ae6e69a19ef0c9e7 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ddbc8d05a54c56891a306e002af685 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03baed5589514993987bedea815156fc : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c154df19f67a48268b3392c52e32b149 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385345ec66c0465c897249ee948f97d2 : SUCCESS in 8m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cfbb05e1f2cb453287730af64f3f7ffa : SUCCESS in 33m 48s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b96b59e1c80548f98e6996cbf6b5fc80 : SUCCESS in 7m 19s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5680745e001f4d8aa2e4a3e62380d8af : SUCCESS in 43m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/086824a751434cadb8cfeed121805489 : SUCCESS in 5m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"511cf74f9756542a1698580615533d36d879d763","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 03:55:12.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1625c3d2daca7e341a1878a7eb29691fa7bec251","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-12-07 05:56:20.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4dbff73d5070f15c529ead4fa1f006938a455ee6","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 15:50:06.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0f21ab937fb14ce9059b94c4ec67e7a3dee7f165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:50:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"95b8fc44febf2df072507544c8e19553cf98b2c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 17:13:08.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19010b5e003f44f2b2cfd6433d9454f0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/440690b78b424fb893ba459f6973d5dc : SUCCESS in 55m 12s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2659ac62a4be4157b17373f41538d258 : SUCCESS in 40m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/74f0e3ec00014631958aa88565542580 : SUCCESS in 38m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fdba4447ad7c48b6a98bc80f6470aa60 : POST_FAILURE in 1h 14m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bf59aff614f34592ac03dc92201b04d1 : SUCCESS in 53m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1d4f279cd1c941d8a8ce43c49920fd51 : SUCCESS in 50m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9172754d8ecc47d7a117d4eb8171aad1 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/837c889684f04598814fdd5083cddb31 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04eaed3b90ce48cdad27e86e0e0f1b75 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06194d86d004221b8236046e197ab6c : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8b006bab61e41e09bcc2cf73f14e678 : SUCCESS in 43m 27s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/b61baffb37c642d19f698c76ca8dbbc4 : SUCCESS in 7m 32s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/5de92cc9a4b74b9f911a466690ff6f43 : SUCCESS in 36m 51s","accounts_in_message":[],"_revision_number":6},{"id":"3e524a720d3e05c42008babb91a805fcf0e9a0ab","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 17:22:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d8655dafed72f1bc184020fd0c6cdec625c3b614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:41:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/546cd6f51fb94e68a1c46a2596ed957f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e68bf4bcbe4d4a3b838a84fbc8af0152 : SUCCESS in 49m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f8f9f0cd31104c3c95ec58e57150aa2f : SUCCESS in 57m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fd89f6396684715a491cf9165ef9705 : SUCCESS in 1h 04m 48s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3e3e5f76d2eb444b87cf98aca14e2062 : SUCCESS in 57m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d33fb270779c412f8675a1b838f0783e : SUCCESS in 48m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/407697967c61438b847385275a699fb4 : SUCCESS in 1h 10m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/79f517aa05fa45649ab7b1eceeb6ca28 : SUCCESS in 1h 02m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9aa681a6ec27418ea04ac9abce309785 : SUCCESS in 41m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4406e2529df5443f9f5ac57dab7c0a31 : SUCCESS in 36m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fde717d5e21748588fa1f9f5698ed8fe : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bb2f0505edb4da0a350c442cbe8b93a : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b402d34de9fe42abb0b63fbc07ea8147 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/660913b33fe148c49e963a109218c512 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23815c8632834154b105a24472b40ad6 : SUCCESS in 7m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a4956c728807483db4d1512d5c903895 : SUCCESS in 45m 54s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/052ad15218b045b5839acb7aea160406 : SUCCESS in 7m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/6ce3e174574c402caa71e8784f16eb7b : SUCCESS in 57m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9ed75dd006344afd9176d25c16c04dad : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2a84db1c4157227c8d3d183ccdd7749125a35f5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:41:59.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"205df0db20b62d9427f3fb166fe0e9c6f699da69","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 18:49:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed6bf52287f0226bc708e22d28c196772d5de42e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 19:53:53.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f4806d8e46743beaf8492abe35a5d71\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/78c1899a90184acf9cf5ee131ca683d0 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a8d0ace5cf554690bc3120310c92546a : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/565c492d6184451380e30e5e63d52e7b : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c5af443b29324bd29ab2dcecf326514f : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/08699144e2664369aacc857477eaabf1 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/46c238a63e1341538c7711f919a163d9 : FAILURE in 37m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8a8e5d1812242b794ffd50b4ccb6b3c : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3755832d656a4281b7869abee006be32 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/746617f31bc94664984362de008a0975 : SUCCESS in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c14bbf1902242f39169275babe21b69 : SUCCESS in 7m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d19e287d36b04dc1a32ebf7a41bce225 : CANCELED\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/312dbd90fc834182997b443d8de3e3fb : SUCCESS in 7m 05s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/4068049ec2f64283a1e6bbe9f5b39d42 : CANCELED","accounts_in_message":[],"_revision_number":6},{"id":"db4c23b8df54c543ac304d1b6b489b071fc1e3d5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 20:06:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d1981e51816384442fc048c4ea98ad0eed2c3d1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:26:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6589c34b90d4ea2b0c37d126db9604c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/863225ad65b04d7d8238eef02a599c8a : SUCCESS in 38m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b20fc60dfa164f649a51020eb7fc9861 : SUCCESS in 48m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e33df27e6e4d46fa8eada1d63d87d8af : SUCCESS in 41m 47s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2d9ea158e1d4471a833fc867d195a002 : SUCCESS in 32m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c9eb7f1a2b6482fad5197535bc6d514 : SUCCESS in 34m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8d93a6fc74af45d59152d38da6dcb6d5 : SUCCESS in 48m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d56d64fe7996417b9bfe16c29a92b025 : SUCCESS in 1h 11m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc3d65305e0b43a898da68fb8dfc590b : SUCCESS in 49m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0caa2c9e9e3d4b53890104e930f34911 : SUCCESS in 35m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d843de7a66d743a5a1c27d4d38ed4a35 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/508ded18fe904f2aa383912144bce10d : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adb5756749ae4abe85f8e7cb2c07011f : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c03a22ace404530bb4df6f52239264e : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2c9a9c61d41494fb66d9d3502bcc970 : SUCCESS in 7m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c2cc7aacf6464a9f8814d56fa2755dbd : SUCCESS in 31m 18s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/e4e8da2b808c40a2a0ebfde90fe1a4a1 : SUCCESS in 7m 08s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ce6c04528ab54a2195242c8ce59aa1ac : SUCCESS in 40m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a646aa985134461984f474969cfbf1e3 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d0a596f471ae0d856ed640fe9df1ccef377e02f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 21:26:45.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c76c131d99ef0b687eeed5825cbb0f41ceb965f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 22:34:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7b75c6e26f242179daba5e241521ae1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/81d18ccc79ad44bc96c449e9a5afe2ba : SUCCESS in 35m 31s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f04bd40abb244e179db4bde90675dafb : SUCCESS in 33m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0be32d3b321746f7a06083f9dd65e843 : SUCCESS in 45m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/74dba3a4be6d4bb8a72524b7c3ea020a : SUCCESS in 49m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/98c7e3f0291d4dc78c82b6fbacce76df : SUCCESS in 32m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/01465ec6d49a447bba0280652457d8ec : SUCCESS in 37m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ebee9febea341d0a0d5ce28767a0652 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2fe0bb00dbe46faa99307b6a2ce5976 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44fd08c13a1d474b9f6c5aeb95c0e421 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c216ee335c724624ae920e46eb5dd284 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/90c7cd8f13d84d6988588ec1918fbd37 : SUCCESS in 34m 11s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/89903c34cf7f4e1898b4a9f93507c2aa : SUCCESS in 7m 36s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/f3ac1db8642848838f78617303d998b7 : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":6},{"id":"6c2fabc836b93de5f7c2835a15e66d17bd941475","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 22:34:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"96cf87c41f20714742726749dcd776df15c3f170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 22:35:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cafeaa756bc04cb2a208b221498602cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d20ac4fe09240e39cc18fbb7635d8e6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b5b9e9902c000f1dd94474e61861956e2f3c2ca2","revisions":{"aa56dbd47cbd90b3ba63d55c9ce51f778c8f030f":{"kind":"REWORK","_number":1,"created":"2023-12-02 15:09:05.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/01/902501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/902501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/902501/1"}}},"commit":{"parents":[{"commit":"1cf3dd75adc72a547b602351ca11c7c25019d70f","subject":"Cleanup context code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf3dd75adc72a547b602351ca11c7c25019d70f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"subject":"Cleaned up rpc loggingutils","message":"Cleaned up rpc loggingutils\n\n- Added basic test coverage.\n- Moved rpc logging to decorators.\n- Enabled DEBUG logging by default for tests.\n- Changed variables names to make them clearer.\n\nChange-Id: Ie7a00355b853a17d06131d46b17003650126c109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa56dbd47cbd90b3ba63d55c9ce51f778c8f030f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/aa56dbd47cbd90b3ba63d55c9ce51f778c8f030f"}]},"branch":"refs/heads/master"},"7d3078097e1bb6e7fbd7573e52e554bbaea69a0c":{"kind":"REWORK","_number":2,"created":"2023-12-02 15:33:06.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/01/902501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/902501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/902501/2"}}},"commit":{"parents":[{"commit":"1cf3dd75adc72a547b602351ca11c7c25019d70f","subject":"Cleanup context code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf3dd75adc72a547b602351ca11c7c25019d70f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:32:56.000000000","tz":-480},"subject":"Cleaned up rpc loggingutils","message":"Cleaned up rpc loggingutils\n\n- Added basic test coverage.\n- Moved rpc logging to decorators.\n- Enabled DEBUG logging by default for tests.\n- Changed variables names to make them clearer.\n\nChange-Id: Ie7a00355b853a17d06131d46b17003650126c109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d3078097e1bb6e7fbd7573e52e554bbaea69a0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7d3078097e1bb6e7fbd7573e52e554bbaea69a0c"}]},"branch":"refs/heads/master"},"bcd9a88aa0a6fb0c59f4f1d29ddf9ec7edd3d3da":{"kind":"REWORK","_number":3,"created":"2023-12-02 15:38:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/01/902501/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/902501/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/902501/3"}}},"commit":{"parents":[{"commit":"1cf3dd75adc72a547b602351ca11c7c25019d70f","subject":"Cleanup context code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf3dd75adc72a547b602351ca11c7c25019d70f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:38:12.000000000","tz":-480},"subject":"Cleaned up rpc loggingutils","message":"Cleaned up rpc loggingutils\n\n- Added basic test coverage.\n- Moved rpc logging to decorators.\n- Enabled DEBUG logging by default for tests.\n- Changed variables names to make them clearer.\n\nChange-Id: Ie7a00355b853a17d06131d46b17003650126c109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bcd9a88aa0a6fb0c59f4f1d29ddf9ec7edd3d3da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bcd9a88aa0a6fb0c59f4f1d29ddf9ec7edd3d3da"}]},"branch":"refs/heads/master"},"e264fda84851146c17db898d825e4c5923b20d39":{"kind":"REWORK","_number":4,"created":"2023-12-02 15:42:41.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/01/902501/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/902501/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/902501/4"}}},"commit":{"parents":[{"commit":"1cf3dd75adc72a547b602351ca11c7c25019d70f","subject":"Cleanup context code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf3dd75adc72a547b602351ca11c7c25019d70f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:42:24.000000000","tz":-480},"subject":"Cleaned up rpc loggingutils","message":"Cleaned up rpc loggingutils\n\n- Added basic test coverage.\n- Moved rpc logging to decorators.\n- Enabled DEBUG logging by default for tests.\n- Changed variables names to make them clearer.\n\nChange-Id: Ie7a00355b853a17d06131d46b17003650126c109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e264fda84851146c17db898d825e4c5923b20d39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e264fda84851146c17db898d825e4c5923b20d39"}]},"branch":"refs/heads/master"},"1cf282f50b81bb095b32303afdafed632ac951ad":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-12-05 11:39:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/01/902501/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/902501/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/902501/5"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 11:39:54.000000000","tz":0},"subject":"Cleaned up rpc loggingutils","message":"Cleaned up rpc loggingutils\n\n- Added basic test coverage.\n- Moved rpc logging to decorators.\n- Enabled DEBUG logging by default for tests.\n- Changed variables names to make them clearer.\n\nChange-Id: Ie7a00355b853a17d06131d46b17003650126c109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf282f50b81bb095b32303afdafed632ac951ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cf282f50b81bb095b32303afdafed632ac951ad"}]},"branch":"refs/heads/master","description":"Rebase"},"b5b9e9902c000f1dd94474e61861956e2f3c2ca2":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-12-06 13:03:54.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/01/902501/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/01/902501/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/01/902501/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/01/902501/6"}}},"commit":{"parents":[{"commit":"4a84b9c0c2b91d154b85b5c5c1a2fff3612949bc","subject":"Merge \"Fix RPC tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4a84b9c0c2b91d154b85b5c5c1a2fff3612949bc"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 15:07:35.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-06 13:03:54.000000000","tz":0},"subject":"Cleaned up rpc loggingutils","message":"Cleaned up rpc loggingutils\n\n- Added basic test coverage.\n- Moved rpc logging to decorators.\n- Enabled DEBUG logging by default for tests.\n- Changed variables names to make them clearer.\n\nChange-Id: Ie7a00355b853a17d06131d46b17003650126c109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5b9e9902c000f1dd94474e61861956e2f3c2ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5b9e9902c000f1dd94474e61861956e2f3c2ca2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
