)]}'
{"id":"openstack%2Fdesignate~902502","triplet_id":"openstack%2Fdesignate~master~Idda68bdb82a79c723b18482974cf65f9807bcf40","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2023-12-07 15:50:32.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Idda68bdb82a79c723b18482974cf65f9807bcf40","subject":"Improved dnsutils coverage","status":"MERGED","created":"2023-12-02 16:16:04.000000000","updated":"2023-12-07 18:23:35.000000000","submitted":"2023-12-07 18:22:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"902502","meta_rev_id":"059be367e1ab577a3f1dae4f91648782678ea061","_number":902502,"virtual_id_number":902502,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-07 18:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-12-07 03:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-12-07 15:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-02 17:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 03:07:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ad9fbdb0922be45a837d94a2abc9539069e16c18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-02 16:16:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1fb9690ef7c9238427e56ab23699161d6944cd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-02 17:11:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31aaf3aa14e84e54a62996bf8103a939\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/53d56f09cd6c4ae89e9c25ba9005819e : SUCCESS in 41m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1142f745ed044c7aa359334b025d38b1 : SUCCESS in 46m 13s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/056665d82d1d40ffa4ea61314b72251f : SUCCESS in 54m 04s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0803db72758440299343126c626e1792 : SUCCESS in 38m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1c16f3948b044f02be9875c5da672d68 : SUCCESS in 47m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/078db0390d48428fad778f90e54625c3 : SUCCESS in 45m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cf40b137eed348bda071f32959a313c4 : SUCCESS in 52m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/70dc9fd107eb4310820dd9fef29ac279 : SUCCESS in 41m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b471affbcfcb4d8481da8c4a931ee3bd : SUCCESS in 39m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c805cc991c6849fa8ae9762149148f17 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c4e3e33276b4b6f86ee9e64535793de : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abfd34cf5de34a6ca2d8a19cf1b2830c : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed2f39a1c83544849420451fd9812a5f : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/819ffe2381184c7cba761f1c209726a4 : SUCCESS in 7m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b65ca9066efa49d3a112750871538ed6 : SUCCESS in 32m 52s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/643bdbba86db4588ba527c4e350a8dbc : SUCCESS in 7m 59s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/bedd46e5a141441db8d08e70eb697bff : SUCCESS in 44m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f4f54594e04d44bd9293b3a4f7dfd005 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c7bb3a1e9b37c539a561eec91fcd8ccbcf4e7b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-05 11:40:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a2bff010528c0f5201c2083a0117fe8c89cb992b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 12:56:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9def8e380d0546c084b8b45fbf7ce22f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/22ee281852b345abaf33ff1b73a0708d : SUCCESS in 46m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e9948d6ff44a4ff98d5cae24ea38d4af : SUCCESS in 49m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d59a3767dd2d4a3ab86c1c12d21d2e35 : SUCCESS in 38m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1ed8cbc4d788419e95d74762760e8ad2 : SUCCESS in 35m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/582171279fa24b09ba4c575eedb022bd : SUCCESS in 50m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4ba32f5fb5114213be434f8bd8bddcb2 : SUCCESS in 1h 07m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5fd859d866e946848b7870c6258077ba : SUCCESS in 1h 10m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3cd3ca4b259a4f3591d260449d734627 : SUCCESS in 49m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0e1e3fdd9d5f45af96ebcddfff2298fc : SUCCESS in 48m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c15a51b1a1f14a7c9a9551ff21a1f069 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e54a1883f7949ae8b44b9b53ec94f32 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8429f8a5d7444168b1927654b41166e0 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5aad06af937848779b8d57a917cda404 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edbcb40111244e6abee4cfc783d8fa1 : SUCCESS in 8m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1cb24d1bf3024b2cadf56d53cda921e7 : SUCCESS in 45m 30s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8951d6b20c3245609e931bce88617d8c : SUCCESS in 6m 22s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/acaab40bebc644258de4a6ef341b44f9 : SUCCESS in 35m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cfa5fe2b038944908fcf2bf4ea69808c : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"692f3fa80f1504fe532c3949f582eeceaecf29d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 03:07:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"acfc693ede95a856a8a67555c3faea16c5d1a80a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 15:50:32.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b83e7fe09e25118beac633f0644b04e1be599be8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2023-12-07 15:50:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c6532e670e7eb68dcefcf9cfabe5375c82ddf8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:51:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9621d10adf2ddedaa3f1c663ab52257772619902","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:22:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0156f2b4117d4a9891a7f732e7a063e5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a425f672b1504e66a412a5d0abd1f411 : SUCCESS in 40m 10s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7366948ec5324e17ab3f34a1703bd2a7 : SUCCESS in 40m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/77cf013a679d484eb98340e1d39c3f24 : SUCCESS in 47m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a41dd4e14b3e4cd2914d5e49a87c8f21 : SUCCESS in 1h 04m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cd85b0b532144db7aefcc693fa479ea6 : SUCCESS in 40m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ec88a92b6f124660be193c66cbda1a12 : SUCCESS in 53m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb698416206e44aea4135fddba7e8690 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69ffae30db8741acb7311e4fd3185538 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d247bed75552432ca99680f44a4de55f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc576281dbe4531afeebf6ad8c45022 : SUCCESS in 7m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71bb9cd809b34a7e81b92b61d7f325ad : SUCCESS in 40m 30s\n- designate-tox-py310-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/ee5c4ee70057459ca0377764a677e466 : SUCCESS in 6m 54s\n- designate-bind9-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/2022efba44094b16af656a90c9096a19 : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":2},{"id":"627e43567a4a71dfc62b8c789e63f966efb706bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:22:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"059be367e1ab577a3f1dae4f91648782678ea061","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 18:23:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ae945abf0543f3835f031805fa410e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d420480c5ca4002b6fffbbc2a5739ad : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff498b855ee42fca62df38f733dfe88a135d2429","revisions":{"19769aef5b15d094e24a43d403cdbe6e7bc35516":{"kind":"REWORK","_number":1,"created":"2023-12-02 16:16:04.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/02/902502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/02/902502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/902502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/902502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/902502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/02/902502/1"}}},"commit":{"parents":[{"commit":"e264fda84851146c17db898d825e4c5923b20d39","subject":"Cleaned up rpc loggingutils","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e264fda84851146c17db898d825e4c5923b20d39"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 16:15:38.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 16:15:38.000000000","tz":-480},"subject":"Improved dnsutils coverage","message":"Improved dnsutils coverage\n\n- Removed impossible None check.\n\nChange-Id: Idda68bdb82a79c723b18482974cf65f9807bcf40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19769aef5b15d094e24a43d403cdbe6e7bc35516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19769aef5b15d094e24a43d403cdbe6e7bc35516"}]},"branch":"refs/heads/master"},"ff498b855ee42fca62df38f733dfe88a135d2429":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-12-05 11:40:18.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/02/902502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/02/902502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/902502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/902502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/02/902502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/02/902502/2"}}},"commit":{"parents":[{"commit":"98f1ad2856a0acb9081c5449ab13718714b42920","subject":"Merge \"Improved MDNS Handler test coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/98f1ad2856a0acb9081c5449ab13718714b42920"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-02 16:15:38.000000000","tz":-480},"committer":{"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","date":"2023-12-05 11:40:18.000000000","tz":0},"subject":"Improved dnsutils coverage","message":"Improved dnsutils coverage\n\n- Removed impossible None check.\n\nChange-Id: Idda68bdb82a79c723b18482974cf65f9807bcf40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff498b855ee42fca62df38f733dfe88a135d2429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff498b855ee42fca62df38f733dfe88a135d2429"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
